)]}'
{"id":"openstack%2Fproject-config~978566","triplet_id":"openstack%2Fproject-config~master~I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-04-14 18:47:08.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-15 14:18:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491","subject":"propose-updates: Add pcu target","status":"MERGED","created":"2026-03-02 16:52:45.000000000","updated":"2026-04-15 14:21:56.000000000","submitted":"2026-04-15 14:18:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978566","meta_rev_id":"76823d5f7236092d46bd32761767c9a6b80b5ba0","_number":978566,"virtual_id_number":978566,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-15 14:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-04-14 18:47:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-15 14:06:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-15 14:06:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-02 17:43:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-02 18:14:21.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2026-03-02 18:19:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2026-03-03 07:32:23.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-04-15 14:06:45.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"1384cadf6c66f7432883e8a2108b254de358fb2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-02 16:52:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fb8ef58f113ffbd0a280df0880028803770619cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 17:43:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b480b4e0f26f4e63b2b8895792800de9\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e3cfbd5fc3134709ae62ee1e6e0e3591 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":1},{"id":"192a6531ec9acf22fca8f8e56667c76af6c14fd0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-03-02 18:14:21.000000000","message":"Patch Set 1:\n\nPresumably projects relying on this workflow would limit the updater to only pin entries which are blocked from inclusion in global upper constraints?","accounts_in_message":[],"_revision_number":1},{"id":"68af05c1ff748c2b4d9cfbba50891d6e9bbc7d84","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-02 19:11:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bf5e5cefed91b9729595a3e9029a012fef04086","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-03 07:32:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e57c8c89a610aac06c31941b12da432e6229aa47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-03 17:44:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c216c11733e4669598c9408cfe53ae2b1a7fa897","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-03 17:44:38.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"d6db3fa2b2b903ee868c9faff5e63e6f1ddaf832","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-03 17:48:48.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0d26fcffe421493e5acf698b5e99ff176be8f375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 17:52:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ea5322aa6a647c6b626cb7677743d12\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dfcdf8cdcf554978b8b34091bebf1030 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":3},{"id":"de29bb9697d332d084b4ac3a76b5e059b32ba890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-03 19:01:22.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b7c4f98cf3044ca6137631ec3c33aedb4a94e608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 19:11:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ec777459a1a4e9b985186ca525440d4\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d0307cad22b5463e965d74362674de6d : FAILURE in 4m 08s","accounts_in_message":[],"_revision_number":4},{"id":"6aa9af769b24165c8f012d21653ee2132755d410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-03 19:30:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"f744052886fa816d614074192e0a1765b32de9e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 19:42:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1455a7cc4fd04c7bab4511e453b16349\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/be3429dd79c241d9b0bacc3faf3a9726 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":5},{"id":"a1ba5321208cd09681b8edacfe38cd02592a2066","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-03-30 12:44:40.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4b60ac6dd2df32b656d25938eab368d099e20646","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-07 07:58:41.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"5a1c6cf70c416293806e938d24155e766fbc769e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-07 07:58:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2b1961e673cd4cd54f56f626869bee063388ea2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-07 08:11:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65875090ba0f4360853fc405239c6d97\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/47d4ef8d7eb8421c8d8012612260c9d9 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":6},{"id":"331c1bc150f99ae89c5f4fc783bfe4cef43600ff","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-13 12:39:37.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"da6870f9b26d6d174d5cbb98ac7bed8cac83d34f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-13 12:39:49.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"f11b4587e601e31333a996cdd8c6957617314f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 12:46:51.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3d04439e9d443209b191757649891f2\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/b1fd6fc8e3954868a4ddef6219b49aa8 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":7},{"id":"c6b4179cd84498f9da58b8b1767be5fd0ae08f0f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-13 12:56:39.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"526dedb8294affd5f55fd83ae68c00a6224441e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-13 12:58:59.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"fbcfeff5d4e099eb5d465f29b03ed0fb94138cd8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-13 12:59:04.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6f4d95b74f565b26d1b944c64d2e732229d51501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-13 13:06:50.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"29ef84b26ab18349ddcbfcfbcb431935f24461c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 13:21:37.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a83723758c74526a8201d48345146d9\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/398405af2066426db72ee6d6b7afad64 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":9},{"id":"943a085374b8554f0231c16857d0c1a143a29101","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-14 07:55:23.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"23db9c1bd962bd0759e5735f030b50774faa8f8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-14 08:10:00.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":10},{"id":"daec19ac365403984568c5645f459f9e04a00db5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-14 08:10:07.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ec0f11bef24c0db42d3962f6ec394ef8e85b800a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 08:21:00.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/443eea70a2a840968dd4e65ef88d681e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8cb32d86bbfd438bb4d70dfc1f9a415d : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":10},{"id":"9b1dd52ae459e407461225ea63c9d8a76bd0d530","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-04-14 18:47:08.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"57139e786fa379cd656db15f6f624464bc3de239","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-04-15 14:06:45.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nLet\u0027s give it a try, we just need to keep an eye out for unexpected failures in other proposal jobs.\n\nI\u0027m personally unconvinced with the tradeoffs in design (the added complexity of a shared venv doesn\u0027t bring much improvement in efficiency, while the proposal jobs will now install this extra tool regardless of whether they need to use it). That said, I don\u0027t feel strongly enough to block this effort.","accounts_in_message":[],"_revision_number":10},{"id":"794f7a231b408f71dfec3a85708960a2b8f21e3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 14:07:24.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"61e8311e366778a38795e456791cc7ecef66b0a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 14:18:45.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02696610c4504498b13fc2c61e173c0e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/eee2f5ef65ac4f7c902acaef4516c4b4 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":10},{"id":"99f3c33623046c82d4e40818d8e5bd8afeac262b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 14:18:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"76823d5f7236092d46bd32761767c9a6b80b5ba0","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 14:21:56.000000000","message":"Patch Set 10:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f890830fe634a61b253e9a0e2637d00\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/0ae8294615f94652b09a9a3c2ff9bf7e : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"5babc52586992fdfe5420023b835b8f12035e59f","revisions":{"e2605e7567a98535a0d8c1a3d42113849465d189":{"kind":"REWORK","_number":1,"created":"2026-03-02 16:52:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/978566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/978566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/978566/1"}}},"commit":{"parents":[{"commit":"61d9c622508d3de7dd46ba04b349fb75aa592ff2","subject":"Add metapackages repo to StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61d9c622508d3de7dd46ba04b349fb75aa592ff2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-02 16:38:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-02 16:52:43.000000000","tz":60},"subject":"propose-updates: Add test-requirements target","message":"propose-updates: Add test-requirements target\n\nThe idea is to propose updates to test-requirements.txt to allow\nprojects to pin them to given minor version (e.g. pin ansible-lint\nto get reproducible linting and handle issues that are outcome\nof a new ansible-lint version in an update patch)\n\nIn order to do that a project needs to have test-reqs-update tox\nenv in their tox.ini and utilise e.g. pur [1] to facilitate the\nupdate.\n\n[1]: https://pypi.org/project/pur/\n\nExample of usage in https://review.opendev.org/c/openstack/kolla/+/978565\n\nChange-Id: I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2605e7567a98535a0d8c1a3d42113849465d189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e2605e7567a98535a0d8c1a3d42113849465d189"}]},"branch":"refs/heads/master"},"8811603a2b8127aed2f4c5efb8bfc88b68117c33":{"kind":"REWORK","_number":2,"created":"2026-03-03 17:44:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/978566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/978566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/978566/2"}}},"commit":{"parents":[{"commit":"61d9c622508d3de7dd46ba04b349fb75aa592ff2","subject":"Add metapackages repo to StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61d9c622508d3de7dd46ba04b349fb75aa592ff2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-02 16:38:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-03 17:44:17.000000000","tz":60},"subject":"propose-updates: Add ansible-lint","message":"propose-updates: Add ansible-lint\n\nKolla (amongst other projects) is using ansible-lint for linting\nAnsible code. A common problem is, that new ansible-lint versions\nare introducing new checks and even pinning to minor versions\n(eg. 2025.12) sometimes introduces new linting failures.\n\nOn other side - updating ansible-lint pins in test-requirements.txt\nis also a manual task, which we\u0027d like to avoid.\n\nThis patch facilitates pip-check-updates [1] which looks actively\nmaintained.\n\n[1]: https://github.com/zehengl/pip-check-updates\n\nChange-Id: I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8811603a2b8127aed2f4c5efb8bfc88b68117c33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8811603a2b8127aed2f4c5efb8bfc88b68117c33"}]},"branch":"refs/heads/master"},"60fb27b64ef12a348a360c32851b99dba277dd91":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-03-03 17:44:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/978566/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/978566/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/978566/3"}}},"commit":{"parents":[{"commit":"61d9c622508d3de7dd46ba04b349fb75aa592ff2","subject":"Add metapackages repo to StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61d9c622508d3de7dd46ba04b349fb75aa592ff2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-02 16:38:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-03 17:44:29.000000000","tz":60},"subject":"propose-updates: Add ansible-lint target","message":"propose-updates: Add ansible-lint target\n\nKolla (amongst other projects) is using ansible-lint for linting\nAnsible code. A common problem is, that new ansible-lint versions\nare introducing new checks and even pinning to minor versions\n(eg. 2025.12) sometimes introduces new linting failures.\n\nOn other side - updating ansible-lint pins in test-requirements.txt\nis also a manual task, which we\u0027d like to avoid.\n\nThis patch facilitates pip-check-updates [1] which looks actively\nmaintained.\n\n[1]: https://github.com/zehengl/pip-check-updates\n\nChange-Id: I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60fb27b64ef12a348a360c32851b99dba277dd91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/60fb27b64ef12a348a360c32851b99dba277dd91"}]},"branch":"refs/heads/master"},"82c48b278b77cc5fd9476ebabe7aefef171dad65":{"kind":"REWORK","_number":4,"created":"2026-03-03 19:01:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/978566/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/978566/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/978566/4"}}},"commit":{"parents":[{"commit":"61d9c622508d3de7dd46ba04b349fb75aa592ff2","subject":"Add metapackages repo to StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61d9c622508d3de7dd46ba04b349fb75aa592ff2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-02 16:38:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-03 18:59:30.000000000","tz":60},"subject":"propose-updates: Add pcu target","message":"propose-updates: Add pcu target\n\nKolla (amongst other projects) is using ansible-lint for linting\nAnsible code. A common problem is, that new ansible-lint versions\nare introducing new checks and even pinning to minor versions\n(eg. 2025.12) sometimes introduces new linting failures.\n\nOn other side - updating ansible-lint pins in test-requirements.txt\nis also a manual task, which we\u0027d like to avoid.\n\nThis patch facilitates pip-check-updates [1] which looks actively\nmaintained to create a pcu target, which Kolla and other projects\ncan utilise for updating non-requirements managed tools.\n\nThere is no safeguard for checking if updated packages are not\npart of requirements repository management, but requirements-check\njob in respective project repository should fail on such occasion.\n\n[1]: https://github.com/zehengl/pip-check-updates\n\nChange-Id: I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82c48b278b77cc5fd9476ebabe7aefef171dad65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/82c48b278b77cc5fd9476ebabe7aefef171dad65"}]},"branch":"refs/heads/master"},"29bc68420c9068a2935d22f724f6a0b3a0fc415c":{"kind":"REWORK","_number":5,"created":"2026-03-03 19:30:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/978566/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/978566/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/978566/5"}}},"commit":{"parents":[{"commit":"61d9c622508d3de7dd46ba04b349fb75aa592ff2","subject":"Add metapackages repo to StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61d9c622508d3de7dd46ba04b349fb75aa592ff2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-02 16:38:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-03 19:30:42.000000000","tz":60},"subject":"propose-updates: Add pcu target","message":"propose-updates: Add pcu target\n\nKolla (amongst other projects) is using ansible-lint for linting\nAnsible code. A common problem is, that new ansible-lint versions\nare introducing new checks and even pinning to minor versions\n(eg. 2025.12) sometimes introduces new linting failures.\n\nOn other side - updating ansible-lint pins in test-requirements.txt\nis also a manual task, which we\u0027d like to avoid.\n\nThis patch facilitates pip-check-updates [1] which looks actively\nmaintained to create a pcu target, which Kolla and other projects\ncan utilise for updating non-requirements managed tools.\n\nThere is no safeguard for checking if updated packages are not\npart of requirements repository management, but requirements-check\njob in respective project repository should fail on such occasion.\n\n[1]: https://github.com/zehengl/pip-check-updates\n\nChange-Id: I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/29bc68420c9068a2935d22f724f6a0b3a0fc415c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/29bc68420c9068a2935d22f724f6a0b3a0fc415c"}]},"branch":"refs/heads/master"},"a50fc784c978f8451f02f9b6e7732c95f4d38abd":{"kind":"REWORK","_number":6,"created":"2026-04-07 07:58:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/978566/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/978566/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/978566/6"}}},"commit":{"parents":[{"commit":"61d9c622508d3de7dd46ba04b349fb75aa592ff2","subject":"Add metapackages repo to StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61d9c622508d3de7dd46ba04b349fb75aa592ff2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-02 16:38:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-07 07:58:38.000000000","tz":0},"subject":"propose-updates: Add pcu target","message":"propose-updates: Add pcu target\n\nKolla (amongst other projects) is using ansible-lint for linting\nAnsible code. A common problem is, that new ansible-lint versions\nare introducing new checks and even pinning to minor versions\n(eg. 2025.12) sometimes introduces new linting failures.\n\nOn other side - updating ansible-lint pins in test-requirements.txt\nis also a manual task, which we\u0027d like to avoid.\n\nThis patch facilitates pip-check-updates [1] which looks actively\nmaintained to create a pcu target, which Kolla and other projects\ncan utilise for updating non-requirements managed tools.\n\nThere is no safeguard for checking if updated packages are not\npart of requirements repository management, but requirements-check\njob in respective project repository should fail on such occasion.\n\n[1]: https://github.com/zehengl/pip-check-updates\n\nChange-Id: I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a50fc784c978f8451f02f9b6e7732c95f4d38abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a50fc784c978f8451f02f9b6e7732c95f4d38abd"}]},"branch":"refs/heads/master"},"edb97697df266c8522d925a23a228bb79f89597d":{"kind":"REWORK","_number":7,"created":"2026-04-13 12:39:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/978566/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/978566/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/978566/7"}}},"commit":{"parents":[{"commit":"61d9c622508d3de7dd46ba04b349fb75aa592ff2","subject":"Add metapackages repo to StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61d9c622508d3de7dd46ba04b349fb75aa592ff2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-02 16:38:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-13 12:39:46.000000000","tz":120},"subject":"propose-updates: Add pcu target","message":"propose-updates: Add pcu target\n\nKolla (amongst other projects) is using ansible-lint for linting\nAnsible code. A common problem is, that new ansible-lint versions\nare introducing new checks and even pinning to minor versions\n(eg. 2025.12) sometimes introduces new linting failures.\n\nOn other side - updating ansible-lint pins in test-requirements.txt\nis also a manual task, which we\u0027d like to avoid.\n\nThis patch facilitates pip-check-updates [1] which looks actively\nmaintained to create a pcu target, which Kolla and other projects\ncan utilise for updating non-requirements managed tools.\n\nThere is no safeguard for checking if updated packages are not\npart of requirements repository management, but requirements-check\njob in respective project repository should fail on such occasion.\n\n[1]: https://github.com/zehengl/pip-check-updates\n\nChange-Id: I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/edb97697df266c8522d925a23a228bb79f89597d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/edb97697df266c8522d925a23a228bb79f89597d"}]},"branch":"refs/heads/master"},"7e3388203c26ba0a653f4967944580a9d3d17b59":{"kind":"REWORK","_number":8,"created":"2026-04-13 12:58:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/978566/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/978566/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/978566/8"}}},"commit":{"parents":[{"commit":"61d9c622508d3de7dd46ba04b349fb75aa592ff2","subject":"Add metapackages repo to StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61d9c622508d3de7dd46ba04b349fb75aa592ff2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-02 16:38:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-13 12:58:57.000000000","tz":0},"subject":"propose-updates: Add pcu target","message":"propose-updates: Add pcu target\n\nKolla (amongst other projects) is using ansible-lint for linting\nAnsible code. A common problem is, that new ansible-lint versions\nare introducing new checks and even pinning to minor versions\n(eg. 2025.12) sometimes introduces new linting failures.\n\nOn other side - updating ansible-lint pins in test-requirements.txt\nis also a manual task, which we\u0027d like to avoid.\n\nThis patch facilitates pip-check-updates [1] which looks actively\nmaintained to create a pcu target, which Kolla and other projects\ncan utilise for updating non-requirements managed tools.\n\nThere is no safeguard for checking if updated packages are not\npart of requirements repository management, but requirements-check\njob in respective project repository should fail on such occasion.\n\n[1]: https://github.com/zehengl/pip-check-updates\n\nChange-Id: I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e3388203c26ba0a653f4967944580a9d3d17b59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7e3388203c26ba0a653f4967944580a9d3d17b59"}]},"branch":"refs/heads/master"},"2f54d461660f971aa9243c32327dc0275e96325a":{"kind":"REWORK","_number":9,"created":"2026-04-13 13:06:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/978566/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/978566/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/978566/9"}}},"commit":{"parents":[{"commit":"61d9c622508d3de7dd46ba04b349fb75aa592ff2","subject":"Add metapackages repo to StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61d9c622508d3de7dd46ba04b349fb75aa592ff2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-02 16:38:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-13 13:06:44.000000000","tz":120},"subject":"propose-updates: Add pcu target","message":"propose-updates: Add pcu target\n\nKolla (amongst other projects) is using ansible-lint for linting\nAnsible code. A common problem is, that new ansible-lint versions\nare introducing new checks and even pinning to minor versions\n(eg. 2025.12) sometimes introduces new linting failures.\n\nOn other side - updating ansible-lint pins in test-requirements.txt\nis also a manual task, which we\u0027d like to avoid.\n\nThis patch facilitates pip-check-updates [1] which looks actively\nmaintained to create a pcu target, which Kolla and other projects\ncan utilise for updating non-requirements managed tools.\n\nThere is no safeguard for checking if updated packages are not\npart of requirements repository management, but requirements-check\njob in respective project repository should fail on such occasion.\n\n[1]: https://github.com/zehengl/pip-check-updates\n\nChange-Id: I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2f54d461660f971aa9243c32327dc0275e96325a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2f54d461660f971aa9243c32327dc0275e96325a"}]},"branch":"refs/heads/master"},"5babc52586992fdfe5420023b835b8f12035e59f":{"kind":"REWORK","_number":10,"created":"2026-04-14 08:10:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/66/978566/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/66/978566/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/66/978566/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/66/978566/10"}}},"commit":{"parents":[{"commit":"61d9c622508d3de7dd46ba04b349fb75aa592ff2","subject":"Add metapackages repo to StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/61d9c622508d3de7dd46ba04b349fb75aa592ff2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-03-02 16:38:48.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-14 08:10:00.000000000","tz":0},"subject":"propose-updates: Add pcu target","message":"propose-updates: Add pcu target\n\nKolla (amongst other projects) is using ansible-lint for linting\nAnsible code. A common problem is, that new ansible-lint versions\nare introducing new checks and even pinning to minor versions\n(eg. 2025.12) sometimes introduces new linting failures.\n\nOn other side - updating ansible-lint pins in test-requirements.txt\nis also a manual task, which we\u0027d like to avoid.\n\nThis patch facilitates pip-check-updates [1] which looks actively\nmaintained to create a pcu target, which Kolla and other projects\ncan utilise for updating non-requirements managed tools.\n\nThere is no safeguard for checking if updated packages are not\npart of requirements repository management, but requirements-check\njob in respective project repository should fail on such occasion.\n\n[1]: https://github.com/zehengl/pip-check-updates\n\nChange-Id: I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5babc52586992fdfe5420023b835b8f12035e59f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5babc52586992fdfe5420023b835b8f12035e59f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
