)]}'
{"id":"openstack%2Fproject-config~981156","triplet_id":"openstack%2Fproject-config~master~Ia26fedc394b1ff775d04376504b1c3f85509da33","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2026-03-18 17:36:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia26fedc394b1ff775d04376504b1c3f85509da33","subject":"Temporarily add the power to delete a starlingx branch.","status":"MERGED","created":"2026-03-18 16:56:05.000000000","updated":"2026-03-18 17:54:18.000000000","submitted":"2026-03-18 17:36:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981156","meta_rev_id":"8b52e9ab207192b1a9ffa927df8c6fa374c73d31","_number":981156,"virtual_id_number":981156,"owner":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-18 17:36:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2026-03-18 17:00:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-18 17:27:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-18 17:27:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-18 17:00:02.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-03-18 17:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-18 17:27:04.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"a77f543636b4d3ee52214cf5638e17cb57163238","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2026-03-18 16:56:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb45a25288a79e416c2520007a153e94807f81cf","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-03-18 17:00:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b1cf29e5bafd0ac1fcf50ac929d93859fe85a6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:02:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a15d5d8c276f4a84b6b6edb2d647cf20\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/203749db8cf24c48bdee507095da42ec : SUCCESS in 4m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d9db4cd24cb64aafb214c9966408ec7c : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1fc7a15ed6b0e2930da3d8993f177a1cc1f0f282","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-03-18 17:27:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b479c33a9035031da8131696fe583f5c87f4848","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:27:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e64672d289fdd2227eae7e30ca055a015163826","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:36:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3f6492e5b2747549e68f244cb6c8fd5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7452ce3d24bd41f28ccf1747252972ed : SUCCESS in 3m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ddcdd24b8d7d4a8e81f38e4eb7aebda2 : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9c688a7be526954d0208f02be6939700e7a82043","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:36:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8b52e9ab207192b1a9ffa927df8c6fa374c73d31","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:54:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba98be0175604994b807452ad10aec46\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/699abb87265c41a18ee395e1c9e38fd0 : SUCCESS in 7m 16s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/c7d449120cab4fa3ad41be6b6cdd79e8 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf74bca634f20accc1d92b813cc9fd031e72187e","revisions":{"cf74bca634f20accc1d92b813cc9fd031e72187e":{"kind":"REWORK","_number":1,"created":"2026-03-18 16:56:05.000000000","uploader":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"ref":"refs/changes/56/981156/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/56/981156/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/981156/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/981156/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/56/981156/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/56/981156/1"}}},"commit":{"parents":[{"commit":"24d3bb3a332e251a62268286a33d0987250bdf86","subject":"Revert \"Temporarily add the power to delete a starlingx branch.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24d3bb3a332e251a62268286a33d0987250bdf86"}]}],"author":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2026-03-18 16:55:25.000000000","tz":-240},"committer":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2026-03-18 16:55:32.000000000","tz":-240},"subject":"Temporarily add the power to delete a starlingx branch.","message":"Temporarily add the power to delete a starlingx branch.\n\nA branch was recently pushed to imany starlingx gits in error.\nThis change grants the power to delete that branch.\nThis change will be reverted once the cleanup is complete.\n\nSigned-off-by: Scott Little \u003cscott.little@windriver.com\u003e\nChange-Id: Ia26fedc394b1ff775d04376504b1c3f85509da33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cf74bca634f20accc1d92b813cc9fd031e72187e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cf74bca634f20accc1d92b813cc9fd031e72187e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
