)]}'
{"id":"openstack%2Fproject-config~981924","triplet_id":"openstack%2Fproject-config~master~I98d4d02f33f441ba5d143728c2c3b96506075c9e","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-04-10 14:46:50.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-10 15:01:46.000000000","reason":"Change was submitted"},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-04-10 15:01:46.000000000","reason":"Change was submitted"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-04-02 15:25:26.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-04-09 18:23:48.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I98d4d02f33f441ba5d143728c2c3b96506075c9e","subject":"Introduce OpenStack-Ansible Power Reviewers group","status":"MERGED","created":"2026-03-24 16:05:59.000000000","updated":"2026-04-10 15:20:38.000000000","submitted":"2026-04-10 15:01:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981924","meta_rev_id":"aa528176050b46b1d4479fff96b298042ed574fd","_number":981924,"virtual_id_number":981924,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-10 15:01:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2026-04-10 14:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-09 20:05:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-04-09 18:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2026-04-10 14:46:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-24 16:17:10.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2026-03-24 16:22:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-24 17:44:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-04-09 12:16:29.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-04-09 14:07:28.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-04-09 20:05:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-04-09 20:05:35.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"22a167d73388e1f428afea9ce43bf529b23db7ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-24 16:05:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"542911ac3f9b9a9404f8e2d8cced0a8ff2d99c66","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-03-24 16:17:10.000000000","message":"Patch Set 1:\n\nIf it\u0027s open season on bikeshed colors, \"approvers\" also seems misleading since they can\u0027t Workflow+1 (approving the change to enter the gate pipeline).","accounts_in_message":[],"_revision_number":1},{"id":"21be7457487914e878bbe2927b3153139d389269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 16:22:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e64fd16e9664010b6a3543777efbf14\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/77ad8c2486504fb883dc7f36b08cdf57 : SUCCESS in 3m 28s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/29ca068d5d4a4f03921432e0e943fdf7 : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"31a3c0c9d099a93b0d4484aa2be3c20683c9776d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-24 16:56:43.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8628dce1240e7d3054e54b86978385c3a1226cc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-24 16:58:26.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c87be89ff6acbcb61a6c866b26e6eae8a83ea186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 17:08:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1db0ec9a5e48457e86020feb4f2e9722\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/00adac40c652434a8bb0aac25df3d5d3 : SUCCESS in 5m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4f68e7e3889f473baa0bd61ed7877f14 : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":3},{"id":"ef40df994e2175088f49ec5bfa80e004ccb8343c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-03-24 17:43:45.000000000","message":"Patch Set 3:\n\nIt looks like you may need to add \"approvers\" to the special_teams list at https://opendev.org/openstack/project-config/src/commit/cb9cad3/tools/normalize_acl.py#L291-L303 too.","accounts_in_message":[],"_revision_number":3},{"id":"c384342c79f65a0a0a31aaf6f0acde67b311bfaf","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-03-24 17:44:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c1485ffb20a8b8504d718a001602dcb82f7ae20c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 17:20:40.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e703061afec0d6f594131e8cd0e2ff2b78091474","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-31 17:21:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3f365e681bd895ba24cd6be399e29237e8ff19cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:31:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a626dba54fe849d8b63db5dc51eb3c85\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/644b05f19f134e3aa6239e12af009311 : SUCCESS in 5m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/83a1fc378ab24c1fbe8e04ac88ca78dc : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":4},{"id":"a6c9bb44cc35b8770d10b538ad571393104ea72b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-02 15:25:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f2c2c5c81b70d92d0d33243e29c1aa3bdff9506d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-04-02 17:06:21.000000000","message":"Patch Set 4:\n\nI never saw a response to my earlier comment on terminology: specifically that it seems very confusing to me to have an \"approvers\" group that lacks permission to approve changes. The Workflow +1 vote has a description of \"Approved\" yet this new openstack-ansible-approvers members can\u0027t apply it.","accounts_in_message":[],"_revision_number":4},{"id":"a433038125fae2cc542c26e04478182ec301bc37","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-09 11:51:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"50c63cee76ca2cffd16d40d1041a66d0e9a3fd7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-04-09 12:03:57.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"30f2b8bf92ce21d547aaf4ec4985a0e2de23bd2b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 12:06:46.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4bd44bb107242ead20001eae89ecba9e9f3f6035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 12:13:54.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d01719b610a46e3a4e89e1a0d533901\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/83878fc29ea042b1a174d60d0c4bc2ff : SUCCESS in 4m 47s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/13f4ef9282234ee8a94694f2120ce4da : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":5},{"id":"74295de3049f62956b60ed4f729ee0fae32ec0f3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-04-09 18:23:48.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"99a0225c13681869fa01c3b57f87ae68a75b56ea","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-04-09 20:05:35.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"a6bb7926e14bb0d1dea3ece5ff2252b99ac1f8c0","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-04-10 14:46:50.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"49849e118b7f40b8cbfd8668a212963488e9d946","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 14:47:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bad2f1fb630c2274b592d03fcc5d31871e1ec215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:01:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d7966966f7f4e3aa8ef21083f837966\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/2a60380f1978405ab21a8b9f14b86748 : SUCCESS in 5m 18s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/88eb1215ea764087b8b941694d8f267a : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":5},{"id":"eeefbcc94cb2a6dfe60cde0b0d558598fb60bfd7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:01:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"aa528176050b46b1d4479fff96b298042ed574fd","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-10 15:20:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38dd9bf285334e3fa620183a9714fc2b\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/3183a9c328374d88810a043d6496137f : SUCCESS in 8m 38s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/94448834efc0481b95532ea73f2544db : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"01789fb8c76fae1da9d7fa190ff3b9209afd6123","revisions":{"263d8a5fbaaffdf0619ea261f74b59a07f794150":{"kind":"REWORK","_number":1,"created":"2026-03-24 16:05:59.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/981924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/981924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/981924/1"}}},"commit":{"parents":[{"commit":"cb9cad35eba2e91ee6053f402ce89b6d3987e063","subject":"Add wandertracks-rail repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb9cad35eba2e91ee6053f402ce89b6d3987e063"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-24 16:05:52.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-24 16:05:52.000000000","tz":60},"subject":"Introduce OpenStack-Ansible Approvers group","message":"Introduce OpenStack-Ansible Approvers group\n\nThe intention behind this group is alike to `reviewers` group for Ironic\nor Kolla - escalated privileges for memebers of the group to have a\npermission for Code-Review +2 and Backport Candidate +1 labels.\n\nThis intends to serve as an intermediate step between regular reviewers\nand cores.\n\nWe selected not to name the group as `reviewers` as in my opinion this\ndownplays a role of regular gerrit users in ecosystem. And\nwe want to keep the value of +1 Code Review value as much as possible.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/openstack_ansible_meeting/2026/openstack_ansible_meeting.2026-03-24-15.02.html\n\nChange-Id: I98d4d02f33f441ba5d143728c2c3b96506075c9e\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/263d8a5fbaaffdf0619ea261f74b59a07f794150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/263d8a5fbaaffdf0619ea261f74b59a07f794150"}]},"branch":"refs/heads/master"},"e0a65969894216fd5f9d9bcb2e10f9fbb3a46b4e":{"kind":"REWORK","_number":2,"created":"2026-03-24 16:56:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/981924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/981924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/981924/2"}}},"commit":{"parents":[{"commit":"cb9cad35eba2e91ee6053f402ce89b6d3987e063","subject":"Add wandertracks-rail repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb9cad35eba2e91ee6053f402ce89b6d3987e063"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-24 16:05:52.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-24 16:56:42.000000000","tz":0},"subject":"Introduce OpenStack-Ansible Approvers group","message":"Introduce OpenStack-Ansible Approvers group\n\nThe intention behind this group is alike to `reviewers` group for Ironic\nor Kolla - escalated privileges for memebers of the group to have a\npermission for Code-Review +2 and Backport Candidate +1 labels.\n\nThis intends to serve as an intermediate step between regular reviewers\nand cores.\n\nWe selected not to name the group as `reviewers` as in my opinion this\ndownplays a role of regular gerrit users in ecosystem. And\nwe want to keep the value of +1 Code Review value as much as possible.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/openstack_ansible_meeting/2026/openstack_ansible_meeting.2026-03-24-15.02.html\n\nChange-Id: I98d4d02f33f441ba5d143728c2c3b96506075c9e\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0a65969894216fd5f9d9bcb2e10f9fbb3a46b4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0a65969894216fd5f9d9bcb2e10f9fbb3a46b4e"}]},"branch":"refs/heads/master"},"10059e54c7dba6b4f0ab65d7a279c64dd91cc341":{"kind":"REWORK","_number":3,"created":"2026-03-24 16:58:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/981924/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/981924/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/981924/3"}}},"commit":{"parents":[{"commit":"cb9cad35eba2e91ee6053f402ce89b6d3987e063","subject":"Add wandertracks-rail repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb9cad35eba2e91ee6053f402ce89b6d3987e063"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-24 16:05:52.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-24 16:58:26.000000000","tz":0},"subject":"Introduce OpenStack-Ansible Approvers group","message":"Introduce OpenStack-Ansible Approvers group\n\nThe intention behind this group is alike to `reviewers` group for Ironic\nor Kolla - escalated privileges for memebers of the group to have a\npermission for Code-Review +2 and Backport Candidate +1 labels.\n\nThis intends to serve as an intermediate step between regular reviewers\nand cores.\n\nWe selected not to name the group as `reviewers` as in my opinion this\ndownplays a role of regular gerrit users in ecosystem. And\nwe want to keep the value of +1 Code Review value as much as possible.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/openstack_ansible_meeting/2026/openstack_ansible_meeting.2026-03-24-15.02.html\n\nChange-Id: I98d4d02f33f441ba5d143728c2c3b96506075c9e\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10059e54c7dba6b4f0ab65d7a279c64dd91cc341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/10059e54c7dba6b4f0ab65d7a279c64dd91cc341"}]},"branch":"refs/heads/master"},"0b1476dc03421ebd95eb180acf16edbe54ba5250":{"kind":"REWORK","_number":4,"created":"2026-03-31 17:20:40.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/981924/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/981924/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/981924/4"}}},"commit":{"parents":[{"commit":"cb9cad35eba2e91ee6053f402ce89b6d3987e063","subject":"Add wandertracks-rail repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb9cad35eba2e91ee6053f402ce89b6d3987e063"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-24 16:05:52.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-31 17:20:40.000000000","tz":0},"subject":"Introduce OpenStack-Ansible Approvers group","message":"Introduce OpenStack-Ansible Approvers group\n\nThe intention behind this group is alike to `reviewers` group for Ironic\nor Kolla - escalated privileges for memebers of the group to have a\npermission for Code-Review +2 and Backport Candidate +1 labels.\n\nThis intends to serve as an intermediate step between regular reviewers\nand cores.\n\nWe selected not to name the group as `reviewers` as in my opinion this\ndownplays a role of regular gerrit users in ecosystem. And\nwe want to keep the value of +1 Code Review value as much as possible.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/openstack_ansible_meeting/2026/openstack_ansible_meeting.2026-03-24-15.02.html\n\nChange-Id: I98d4d02f33f441ba5d143728c2c3b96506075c9e\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0b1476dc03421ebd95eb180acf16edbe54ba5250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/0b1476dc03421ebd95eb180acf16edbe54ba5250"}]},"branch":"refs/heads/master"},"01789fb8c76fae1da9d7fa190ff3b9209afd6123":{"kind":"REWORK","_number":5,"created":"2026-04-09 12:03:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/24/981924/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/24/981924/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/24/981924/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/24/981924/5"}}},"commit":{"parents":[{"commit":"cb9cad35eba2e91ee6053f402ce89b6d3987e063","subject":"Add wandertracks-rail repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb9cad35eba2e91ee6053f402ce89b6d3987e063"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-24 16:05:52.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-04-09 12:03:56.000000000","tz":0},"subject":"Introduce OpenStack-Ansible Power Reviewers group","message":"Introduce OpenStack-Ansible Power Reviewers group\n\nThe intention behind this group is alike to `reviewers` group for Ironic\nor Kolla - escalated privileges for memebers of the group to have a\npermission for Code-Review +2 and Backport Candidate +1 labels.\n\nThis intends to serve as an intermediate step between regular reviewers\nand cores.\n\nThis change has been agreed on the regular IRC meeting [1]\n\n[1] https://meetings.opendev.org/meetings/openstack_ansible_meeting/2026/openstack_ansible_meeting.2026-03-24-15.02.html\n\nChange-Id: I98d4d02f33f441ba5d143728c2c3b96506075c9e\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01789fb8c76fae1da9d7fa190ff3b9209afd6123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/01789fb8c76fae1da9d7fa190ff3b9209afd6123"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
