)]}'
{"id":"openstack%2Fproject-config~981962","triplet_id":"openstack%2Fproject-config~master~I94b50669e7c66d92fe24d77920a7f27e365accca","project":"openstack/project-config","branch":"master","topic":"retire-venus","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-09-16 10:26:33.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-09-16 10:16:12.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-16 10:26:33.000000000","reason":"Change was submitted"}},"hashtags":["retire-venus","venus-retirement"],"change_id":"I94b50669e7c66d92fe24d77920a7f27e365accca","subject":"Retire Venus Project","status":"MERGED","created":"2026-03-24 18:16:33.000000000","updated":"2026-09-16 10:38:38.000000000","submitted":"2026-09-16 10:26:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"981962-retire-venus","meta_rev_id":"9da3d45d9e706308b7d59fc47eda668986ef0e8c","_number":981962,"virtual_id_number":981962,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-16 10:26:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-06-17 18:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2026-09-16 10:16:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-09-16 10:16:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-24 18:28:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-05 17:23:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2026-06-17 18:59:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d87c774b743ba8d5f88d849c1a9fa941d4057e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-03-24 18:16:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a5dbc6236fca59a04a23b9727f18e2523d08b4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 18:28:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0c7c9cb0bd342f5997b373bda8cba55\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9015ee2e37b9420499f89ce11e126c20 : SUCCESS in 3m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8b67c5e9d8fa4bc4af324415e58a34a0 : SUCCESS in 3m 51s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/168d5cf454a3431492c73ac7a18e29f0 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"52f0672364fd7e13c958e955b1bf04ff7aed53a8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-05 17:23:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc6f87e1a76d5169c43dbe8decfe65a27d81f368","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2026-06-04 18:04:27.000000000","message":"Hashtags added: retire-venus, venus-retirement","accounts_in_message":[],"_revision_number":1},{"id":"133baeb4ec2ba50a89d85c030f853c84fedb16f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-06-05 13:15:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c3fc6164cef6aa686169b3c59e420721cf7fee62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 13:27:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9acb1751b4f44b428da67ebb1e9eb6ea\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9d19c47df5bc4ada8ec399d38afe001f : SUCCESS in 4m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f14b28dbb1fd4aed90b63e97c0523fe1 : SUCCESS in 4m 59s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/7e195a7fafe245d5a8557db645fd0503 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d2be81761d20957c977b9a265d76e0d881e6fbae","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-17 18:59:46.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c25bade895cd6ca06df8be600d861ec27c0a9459","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-09-16 10:16:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2a6f17b50dea1d8e81366aa8a358222eb265f812","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 10:16:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f6635dfcb8ca33c56487f8fd9909cdbe586f120a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 10:26:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a389bbc4c854294b6fc36bc897686e7\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/3a04d2cf23484ef9b44f746ec7e4bf27 : SUCCESS in 4m 34s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/9d73226e36f64a86af69de668a711eda : SUCCESS in 3m 22s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/72ac4577a85a46f88c9db404608469f9 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"2ea2525d24aaee7e573c8e6f75e0106348caf3f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 10:26:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9da3d45d9e706308b7d59fc47eda668986ef0e8c","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 10:38:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dab3e0a69e7b435e9a17ee635956b874\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/c9f7092f288d4ced8b0a42d93c924f31 : SUCCESS in 10m 54s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/a506f342c7b84254925ce10c10110074 : SUCCESS in 4m 02s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/1cfc51731aaf40edb7aa739a4bf0db56 : SUCCESS in 4m 53s\n- infra-prod-run-accessbot https://zuul.opendev.org/t/openstack/build/1de23bed8b824672b5d224e65edc58c2 : SUCCESS in 8m 42s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/0e463f6240074a6f99fecd01b86ee20b : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"101db889562e75f8e80c07641e0acd99d9e872c4","revisions":{"7d8d1eeae3d2739089e323131bc75f8b43ee2eea":{"kind":"REWORK","_number":1,"created":"2026-03-24 18:16:33.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/981962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/981962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/981962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/981962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/981962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/981962/1"}}},"commit":{"parents":[{"commit":"f31f40d66adc9be8efeeb8144f66819c029242ce","subject":"Stop Venus gating","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/f31f40d66adc9be8efeeb8144f66819c029242ce"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-24 18:16:11.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-24 18:16:11.000000000","tz":60},"subject":"Retire Venus Project","message":"Retire Venus Project\n\nThis completes the project clean-up after its retirement\n\nDepends-On: https://review.opendev.org/c/openstack/governance/+/981959\nChange-Id: I94b50669e7c66d92fe24d77920a7f27e365accca\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d8d1eeae3d2739089e323131bc75f8b43ee2eea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7d8d1eeae3d2739089e323131bc75f8b43ee2eea"}]},"branch":"refs/heads/master"},"101db889562e75f8e80c07641e0acd99d9e872c4":{"kind":"REWORK","_number":2,"created":"2026-06-05 13:15:06.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/62/981962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/62/981962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/981962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/981962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/62/981962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/62/981962/2"}}},"commit":{"parents":[{"commit":"ac9ab14d543a3a7ab61640b2eff4f462817ac1f6","subject":"Normalize projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac9ab14d543a3a7ab61640b2eff4f462817ac1f6"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-03-24 18:16:11.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-06-05 13:14:57.000000000","tz":120},"subject":"Retire Venus Project","message":"Retire Venus Project\n\nThis completes the project clean-up after its retirement\n\nDepends-On: https://review.opendev.org/c/openstack/governance/+/981959\nChange-Id: I94b50669e7c66d92fe24d77920a7f27e365accca\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/101db889562e75f8e80c07641e0acd99d9e872c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/101db889562e75f8e80c07641e0acd99d9e872c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
