)]}'
{"id":"openstack%2Fproject-config~982495","triplet_id":"openstack%2Fproject-config~master~I91293d1d21cb72aecd8f95d31043ee402de8d9a1","project":"openstack/project-config","branch":"master","topic":"starlingx-app-prometheus","attention_set":{},"removed_from_attention_set":{"38864":{"account":{"_account_id":38864,"name":"Sridhar Kumar Ettkepalli","display_name":"Sridhar","email":"sridharkumar.ettkepalli@windriver.com","username":"settkepa"},"last_update":"2026-03-30 14:07:42.000000000","reason":"\u003cGERRIT_ACCOUNT_38864\u003e replied on the change","reason_account":{"_account_id":38864,"name":"Sridhar Kumar Ettkepalli","display_name":"Sridhar","email":"sridharkumar.ettkepalli@windriver.com","username":"settkepa"}},"6547":{"account":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"last_update":"2026-03-30 19:12:10.000000000","reason":"Change was submitted"},"37909":{"account":{"_account_id":37909,"name":"Brad Borgald","email":"brad.borgald@windriver.com","username":"bborgald","status":"Wind River"},"last_update":"2026-03-30 19:12:10.000000000","reason":"Change was submitted"},"37078":{"account":{"_account_id":37078,"name":"Daniel Marques Caires","display_name":"Daniel Caires","email":"DanielMarques.Caires@windriver.com","username":"dcaires"},"last_update":"2026-03-30 14:36:25.000000000","reason":"\u003cGERRIT_ACCOUNT_37078\u003e replied on the change","reason_account":{"_account_id":37078,"name":"Daniel Marques Caires","display_name":"Daniel Caires","email":"DanielMarques.Caires@windriver.com","username":"dcaires"}},"33594":{"account":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"last_update":"2026-03-30 19:12:10.000000000","reason":"Change was submitted"},"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2026-03-30 16:23:46.000000000","reason":"\u003cGERRIT_ACCOUNT_28410\u003e replied on the change","reason_account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-03-30 19:05:34.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I91293d1d21cb72aecd8f95d31043ee402de8d9a1","subject":"Add repo app-prometheus to StarlingX","status":"MERGED","created":"2026-03-27 19:49:15.000000000","updated":"2026-03-30 19:31:54.000000000","submitted":"2026-03-30 19:12:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982495-starlingx-app-prometheus","meta_rev_id":"e37a60343c951851b4958debbb2c9697c2b3fc05","_number":982495,"virtual_id_number":982495,"owner":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":38864,"name":"Sridhar Kumar Ettkepalli","display_name":"Sridhar","email":"sridharkumar.ettkepalli@windriver.com","username":"settkepa"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":37078,"name":"Daniel Marques Caires","display_name":"Daniel Caires","email":"DanielMarques.Caires@windriver.com","username":"dcaires"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-30 19:12:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"recommended":{"_account_id":38864,"name":"Sridhar Kumar Ettkepalli","display_name":"Sridhar","email":"sridharkumar.ettkepalli@windriver.com","username":"settkepa"},"all":[{"value":1,"date":"2026-03-30 14:07:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":38864,"name":"Sridhar Kumar Ettkepalli","display_name":"Sridhar","email":"sridharkumar.ettkepalli@windriver.com","username":"settkepa"},{"value":2,"date":"2026-03-30 16:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2026-03-30 14:36:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37078,"name":"Daniel Marques Caires","display_name":"Daniel Caires","email":"DanielMarques.Caires@windriver.com","username":"dcaires"},{"value":1,"date":"2026-03-30 16:23:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-03-30 19:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":38864,"name":"Sridhar Kumar Ettkepalli","display_name":"Sridhar","email":"sridharkumar.ettkepalli@windriver.com","username":"settkepa"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":37078,"name":"Daniel Marques Caires","display_name":"Daniel Caires","email":"DanielMarques.Caires@windriver.com","username":"dcaires"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-30 19:05:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":37909,"name":"Brad Borgald","email":"brad.borgald@windriver.com","username":"bborgald","status":"Wind River"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"_account_id":37078,"name":"Daniel Marques Caires","display_name":"Daniel Caires","email":"DanielMarques.Caires@windriver.com","username":"dcaires"},{"_account_id":37909,"name":"Brad Borgald","email":"brad.borgald@windriver.com","username":"bborgald","status":"Wind River"},{"_account_id":38864,"name":"Sridhar Kumar Ettkepalli","display_name":"Sridhar","email":"sridharkumar.ettkepalli@windriver.com","username":"settkepa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-27 19:59:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-30 13:46:34.000000000","updated_by":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"reviewer":{"_account_id":37909,"name":"Brad Borgald","email":"brad.borgald@windriver.com","username":"bborgald","status":"Wind River"},"state":"REVIEWER"},{"updated":"2026-03-30 13:46:34.000000000","updated_by":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"},{"updated":"2026-03-30 13:47:03.000000000","updated_by":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"reviewer":{"_account_id":38864,"name":"Sridhar Kumar Ettkepalli","display_name":"Sridhar","email":"sridharkumar.ettkepalli@windriver.com","username":"settkepa"},"state":"REVIEWER"},{"updated":"2026-03-30 13:47:03.000000000","updated_by":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"reviewer":{"_account_id":37078,"name":"Daniel Marques Caires","display_name":"Daniel Caires","email":"DanielMarques.Caires@windriver.com","username":"dcaires"},"state":"REVIEWER"},{"updated":"2026-03-30 16:28:37.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-03-30 16:42:31.000000000","updated_by":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-03-30 16:47:27.000000000","updated_by":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"}],"messages":[{"id":"4cfc1c00013fcd008bf4f3fef640db61ff7b77bc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"date":"2026-03-27 19:49:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b30941447da0799a20fe047b013f78047c8642d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-27 19:59:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1fa5be4799364ff28511b5d9f4b16e4f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/45585dc5d33547b3b8d58a5f32587187 : SUCCESS in 5m 02s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/41e14ccfad764711822c6f39de9b18d2 : FAILURE in 2m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8d99859d2d015454571d7823c0b9ac57e5717832","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"date":"2026-03-30 13:11:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2ee9f26fffa4e0182d72dd2a4cab730eff84ee68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 13:18:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2cf41436752f4001b6e1e7a621bdbeb5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f1178cbbcc8c48aeaa79691b15e8abdc : SUCCESS in 2m 29s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e6fef445938d449ab3f1a3740dd9060f : FAILURE in 5m 47s","accounts_in_message":[],"_revision_number":2},{"id":"318049463b7e279e01e12886dd1fac34af4c2523","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"date":"2026-03-30 13:22:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cb1df1d2e70b431d13d6d7c0066d92ab77b461c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 13:39:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83bf668d99554418aaa0ea87ddd057f0\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d193bc13a820473b85e2c6eee379af15 : SUCCESS in 5m 01s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/546ae8db940844e6b95756f8665f9584 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":3},{"id":"b916c47c8fd530a78a1c6b021a02608ad5f3e270","author":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"date":"2026-03-30 13:46:34.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"5935f70d3d94ff7a89d9259fe139b5fcf1db889b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"date":"2026-03-30 14:03:56.000000000","message":"Topic set to starlingx-app-prometheus","accounts_in_message":[],"_revision_number":3},{"id":"e231612a6ffadcb14b86f5e48aac18066a154d98","author":{"_account_id":38864,"name":"Sridhar Kumar Ettkepalli","display_name":"Sridhar","email":"sridharkumar.ettkepalli@windriver.com","username":"settkepa"},"date":"2026-03-30 14:07:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"71d5eae85e8119c210286d7ecdee844d2498680a","author":{"_account_id":37078,"name":"Daniel Marques Caires","display_name":"Daniel Caires","email":"DanielMarques.Caires@windriver.com","username":"dcaires"},"date":"2026-03-30 14:36:25.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1353ba21e24a7808ae82de8d930908fafc0e7066","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2026-03-30 16:23:46.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"154fec701b28d398f972797072b2738ceed4b03a","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-03-30 16:28:37.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"584ce0a7890a8256a59f841c1213617eaea94604","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-03-30 19:05:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"0519531b4718a09980e73faf562a236354427a76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 19:05:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9d633ebc01306c3dfaa0a9d0fbe9fabcd722846d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 19:12:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4ab08bc5fa447aea8e9cf5930046f10\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ba4e8f790c954ae5b00d8de6694ed7d2 : SUCCESS in 4m 23s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5883d03742b648f38ee30efe60bd6ea7 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":3},{"id":"a919034c2e4d0205e5fd0e1698ba1b86ddc8253b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 19:12:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a81f7b767c28c70fb2df44d7efe98161c8fa86cb","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 19:30:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f96cb9d890124a2ca70344e4c742f45e\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/64ec6d25751f4c00a682fe7f20de3316 : SUCCESS in 18m 31s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/e0400f9cdc694ac2ad54b896b8298806 : SUCCESS in 10m 52s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/37a7ceeb20b94a66a0442e84100d1c84 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3},{"id":"e37a60343c951851b4958debbb2c9697c2b3fc05","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-03-30 19:31:54.000000000","message":"Patch Set 3:\n\nI\u0027ve added Scott as the initial member of the starlingx-app-prometheus-core group.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e3f733d369cf1901bafcbec7880fb86edeb0cbc4","revisions":{"8c4304b26fd2b6b7b5ae1d49bc6a25ee78eb31c0":{"kind":"REWORK","_number":1,"created":"2026-03-27 19:49:15.000000000","uploader":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"ref":"refs/changes/95/982495/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/982495/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/982495/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/982495/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/982495/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/982495/1"}}},"commit":{"parents":[{"commit":"cb9cad35eba2e91ee6053f402ce89b6d3987e063","subject":"Add wandertracks-rail repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb9cad35eba2e91ee6053f402ce89b6d3987e063"}]}],"author":{"name":"Thales Elero Cervi","email":"thaleselero.cervi@windriver.com","date":"2026-03-27 19:47:08.000000000","tz":-180},"committer":{"name":"Thales Elero Cervi","email":"thaleselero.cervi@windriver.com","date":"2026-03-27 19:47:08.000000000","tz":-180},"subject":"Add repo app-prometheus to StarlingX","message":"Add repo app-prometheus to StarlingX\n\nThis new repository will be used for source control on the new\nStarlingX application deploying Prometheus to the platform.\n\nSigned-off-by: Thales Elero Cervi \u003cthaleselero.cervi@windriver.com\u003e\nChange-Id: I91293d1d21cb72aecd8f95d31043ee402de8d9a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c4304b26fd2b6b7b5ae1d49bc6a25ee78eb31c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c4304b26fd2b6b7b5ae1d49bc6a25ee78eb31c0"}]},"branch":"refs/heads/master"},"8c70968030807eecee9c1290bb4efe18bd599082":{"kind":"REWORK","_number":2,"created":"2026-03-30 13:11:40.000000000","uploader":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"ref":"refs/changes/95/982495/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/982495/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/982495/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/982495/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/982495/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/982495/2"}}},"commit":{"parents":[{"commit":"cb9cad35eba2e91ee6053f402ce89b6d3987e063","subject":"Add wandertracks-rail repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb9cad35eba2e91ee6053f402ce89b6d3987e063"}]}],"author":{"name":"Thales Elero Cervi","email":"thaleselero.cervi@windriver.com","date":"2026-03-27 19:47:08.000000000","tz":-180},"committer":{"name":"Thales Elero Cervi","email":"thaleselero.cervi@windriver.com","date":"2026-03-30 13:11:31.000000000","tz":-180},"subject":"Add repo app-prometheus to StarlingX","message":"Add repo app-prometheus to StarlingX\n\nThis new repository will be used for source control on the new\nStarlingX application deploying Prometheus to the platform.\n\nSigned-off-by: Thales Elero Cervi \u003cthaleselero.cervi@windriver.com\u003e\nChange-Id: I91293d1d21cb72aecd8f95d31043ee402de8d9a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c70968030807eecee9c1290bb4efe18bd599082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8c70968030807eecee9c1290bb4efe18bd599082"}]},"branch":"refs/heads/master"},"e3f733d369cf1901bafcbec7880fb86edeb0cbc4":{"kind":"REWORK","_number":3,"created":"2026-03-30 13:22:40.000000000","uploader":{"_account_id":33594,"name":"Thales Elero Cervi","display_name":"Thales Cervi","email":"thaleselero.cervi@windriver.com","username":"tcervi"},"ref":"refs/changes/95/982495/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/95/982495/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/982495/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/982495/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/95/982495/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/95/982495/3"}}},"commit":{"parents":[{"commit":"cb9cad35eba2e91ee6053f402ce89b6d3987e063","subject":"Add wandertracks-rail repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cb9cad35eba2e91ee6053f402ce89b6d3987e063"}]}],"author":{"name":"Thales Elero Cervi","email":"thaleselero.cervi@windriver.com","date":"2026-03-27 19:47:08.000000000","tz":-180},"committer":{"name":"Thales Elero Cervi","email":"thaleselero.cervi@windriver.com","date":"2026-03-30 13:22:14.000000000","tz":-180},"subject":"Add repo app-prometheus to StarlingX","message":"Add repo app-prometheus to StarlingX\n\nThis new repository will be used for source control on the new\nStarlingX application deploying Prometheus to the platform.\n\nSigned-off-by: Thales Elero Cervi \u003cthaleselero.cervi@windriver.com\u003e\nChange-Id: I91293d1d21cb72aecd8f95d31043ee402de8d9a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3f733d369cf1901bafcbec7880fb86edeb0cbc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e3f733d369cf1901bafcbec7880fb86edeb0cbc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
