)]}'
{"id":"openstack%2Fproject-config~983475","triplet_id":"openstack%2Fproject-config~master~I172b9fe41dca808e1d74b7d7e1391443e708dbf4","project":"openstack/project-config","branch":"master","topic":"jeepyb-gerrit-3.12","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-04-13 19:33:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I172b9fe41dca808e1d74b7d7e1391443e708dbf4","subject":"Bump jeepyb gerrit jobs up to gerrit 3.12","status":"MERGED","created":"2026-04-06 20:44:50.000000000","updated":"2026-04-13 19:36:20.000000000","submitted":"2026-04-13 19:33:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"983475-jeepyb-gerrit-3.12","meta_rev_id":"4ac673a5480b719e533d72b7b261f5fb77c7731b","_number":983475,"virtual_id_number":983475,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-13 19:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-04-13 19:28:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-04-13 19:28:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-06 20:57:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-13 19:28:21.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"4a62990eb7e2b641f589881f3211339cec52d921","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-04-06 20:44:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75c3877697430a97c485c841a8079b3b270fe299","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-04-06 20:45:07.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9e6e791eb0c0d5209b3a60abf3b71d7a655f4da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-04-06 20:47:27.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"304ccd86299d3706fa8b59b92194f128c0059bc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-06 20:57:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daccec72c6b14afdb9b6672291999975\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1b783fe50cb1434ab3eeac8f06ebda5c : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"5265cf536d572f174807aa1f3a270d146fbd9b6a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-04-13 19:28:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"efc77077a0996f63d10d4eb6fe34a0fa115d74a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 19:28:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9a5c07a9997c23e4970bcf39e458e094f2edb458","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 19:33:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99d7b22db99e4710a612be6a6124d46f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/57d87c338a024f209059c0054c8ecb37 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e2a48ee701ce71f7c0c48949aab7a0b150b996b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 19:33:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4ac673a5480b719e533d72b7b261f5fb77c7731b","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 19:36:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9a830dca2e54dc5a329636a61c763e3\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/2e3a30fa8dca4187a15fd10e59f39fcc : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a0f5748dd6d9260a77814a047c97dd728f40f456","revisions":{"c94d80089adf7c1e32fe76f7f2e29ac4684ce29b":{"kind":"REWORK","_number":1,"created":"2026-04-06 20:44:50.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/75/983475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/983475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/983475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/983475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/983475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/983475/1"}}},"commit":{"parents":[{"commit":"509f313e4e6813c4ed23a1dd090b135427246a9c","subject":"Merge \"Revert \"Temporarily remove release docs semaphores\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/509f313e4e6813c4ed23a1dd090b135427246a9c"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2026-04-06 20:43:48.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2026-04-06 20:43:48.000000000","tz":-420},"subject":"Bump jeepyb gerrit jobs up to gerrit 3.12","message":"Bump jeepyb gerrit jobs up to gerrit 3.12\n\nThis is in preparation for the Gerrit 3.12 upgrade. Currently we run\nGerrit 3.11 and we have jeepyb update that image for us if there are\njeepyb changes. Once we upgrade to Gerrit 3.12 we will want jeepyb to\ntrigger Gerrit 3.12 image builds. This change applies that update and\nshould only be merged after we have upgraded to Gerrit 3.12.\n\nChange-Id: I172b9fe41dca808e1d74b7d7e1391443e708dbf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c94d80089adf7c1e32fe76f7f2e29ac4684ce29b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c94d80089adf7c1e32fe76f7f2e29ac4684ce29b"}]},"branch":"refs/heads/master"},"a0f5748dd6d9260a77814a047c97dd728f40f456":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-04-06 20:47:27.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/75/983475/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/75/983475/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/983475/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/983475/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/75/983475/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/75/983475/2"}}},"commit":{"parents":[{"commit":"509f313e4e6813c4ed23a1dd090b135427246a9c","subject":"Merge \"Revert \"Temporarily remove release docs semaphores\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/509f313e4e6813c4ed23a1dd090b135427246a9c"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2026-04-06 20:43:48.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2026-04-06 20:47:20.000000000","tz":-420},"subject":"Bump jeepyb gerrit jobs up to gerrit 3.12","message":"Bump jeepyb gerrit jobs up to gerrit 3.12\n\nThis is in preparation for the Gerrit 3.12 upgrade. Currently we run\nGerrit 3.11 and we have jeepyb update that image for us if there are\njeepyb changes. Once we upgrade to Gerrit 3.12 we will want jeepyb to\ntrigger Gerrit 3.12 image builds. This change applies that update and\nshould only be merged after we have upgraded to Gerrit 3.12.\n\nChange-Id: I172b9fe41dca808e1d74b7d7e1391443e708dbf4\nSigned-off-by: Clark Boylan \u003cclark.boylan@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a0f5748dd6d9260a77814a047c97dd728f40f456"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a0f5748dd6d9260a77814a047c97dd728f40f456"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
