)]}'
{"id":"openstack%2Fproject-config~984804","triplet_id":"openstack%2Fproject-config~master~I2fca9d4f6012cc9e4a3c38b169d059cde18fceab","project":"openstack/project-config","branch":"master","attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-04-15 15:42:57.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I2fca9d4f6012cc9e4a3c38b169d059cde18fceab","subject":"Install pip-check-updates only in jobs requiring that","status":"NEW","created":"2026-04-15 15:25:36.000000000","updated":"2026-04-16 07:00:13.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2a1f43aeb6864ec29ee0b62d71bf40e8b87a4e0e","_number":984804,"virtual_id_number":984804,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-16 07:00:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-15 15:42:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a5b69dd462386e24f6935507b9c3a2cc75103752","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-15 15:25:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"812a8fb08e27fbfb8293dc4cd568f6c34a6652b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 15:42:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c48ce2ad84914946b53dad6c371a5117\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9acd04a3fe6443b2bfa444c791af0e71 : FAILURE in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e2750257eb45af6789165bd8f6a904176a8f1d09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-15 15:48:19.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"82019636a676bc9b0bbe98edd186ebe35105dc90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-15 16:00:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3749d59da04b4318bf39ad34f79c1415\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1e6815154f2d4a99b68b46080799cabb : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"db4ce511e92c7b44f1d483de02da3aaf1d547eac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-16 06:08:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c42a018ae06a50fa9144c2756f092412dea0cc71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-16 06:09:17.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"d035788073a59d265fed2ad1fb3cd15ac4cdc9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 06:19:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4df955b2c4b442ada9b5703412e0d92c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0447b7a664e44a64b103eb7f3f919bc8 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":4},{"id":"83a3e25a1fc7367959215e6df0b4f86eb875f228","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-04-16 06:49:24.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"2a1f43aeb6864ec29ee0b62d71bf40e8b87a4e0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-16 07:00:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa21902369c14de1bdc5c072bb0ed633\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/43d29753a43a4b75b2c507e8337ee023 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"8fca946ae688994d570148c335bc08d91b2facb1","revisions":{"5cfed4ab146c7314efecc4bda3bbfc20b49edfc2":{"kind":"REWORK","_number":1,"created":"2026-04-15 15:25:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/984804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/984804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/984804/1"}}},"commit":{"parents":[{"commit":"2a799c79c73ffa957717095daa83ecf4519be009","subject":"Follow up for I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a799c79c73ffa957717095daa83ecf4519be009"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-15 15:24:45.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-15 15:24:45.000000000","tz":120},"subject":"Install pip-check-updates only in jobs requiring that","message":"Install pip-check-updates only in jobs requiring that\n\nChange-Id: I2fca9d4f6012cc9e4a3c38b169d059cde18fceab\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5cfed4ab146c7314efecc4bda3bbfc20b49edfc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5cfed4ab146c7314efecc4bda3bbfc20b49edfc2"}]},"branch":"refs/heads/master"},"d551537b1ed3a1216dc2c9471ad080bd0c44756f":{"kind":"REWORK","_number":2,"created":"2026-04-15 15:48:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/984804/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/984804/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/984804/2"}}},"commit":{"parents":[{"commit":"2a799c79c73ffa957717095daa83ecf4519be009","subject":"Follow up for I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2a799c79c73ffa957717095daa83ecf4519be009"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-15 15:24:45.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-15 15:48:19.000000000","tz":0},"subject":"Install pip-check-updates only in jobs requiring that","message":"Install pip-check-updates only in jobs requiring that\n\nChange-Id: I2fca9d4f6012cc9e4a3c38b169d059cde18fceab\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d551537b1ed3a1216dc2c9471ad080bd0c44756f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d551537b1ed3a1216dc2c9471ad080bd0c44756f"}]},"branch":"refs/heads/master"},"84d690337be11bc46a0a79bb7fd1b1d85f0df88f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-04-16 06:08:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/984804/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/984804/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/984804/3"}}},"commit":{"parents":[{"commit":"da4acbba1fb8f37ec33ea489349e4a1169ed5bc1","subject":"Follow up for I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/da4acbba1fb8f37ec33ea489349e4a1169ed5bc1"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-15 15:24:45.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-16 06:08:45.000000000","tz":0},"subject":"Install pip-check-updates only in jobs requiring that","message":"Install pip-check-updates only in jobs requiring that\n\nChange-Id: I2fca9d4f6012cc9e4a3c38b169d059cde18fceab\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/84d690337be11bc46a0a79bb7fd1b1d85f0df88f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/84d690337be11bc46a0a79bb7fd1b1d85f0df88f"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"d551537b1ed3a1216dc2c9471ad080bd0c44756f","theirs":"da4acbba1fb8f37ec33ea489349e4a1169ed5bc1","contains_conflicts":false}},"e0207c86622f4f767f6ef9fdb71b1827b7e4bff1":{"kind":"NO_CHANGE","_number":4,"created":"2026-04-16 06:09:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/984804/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/984804/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/984804/4"}}},"commit":{"parents":[{"commit":"2bc2b6c61b5463794c8cf545fb081a1b34d4ea90","subject":"Follow up for I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2bc2b6c61b5463794c8cf545fb081a1b34d4ea90"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-15 15:24:45.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-16 06:09:17.000000000","tz":0},"subject":"Install pip-check-updates only in jobs requiring that","message":"Install pip-check-updates only in jobs requiring that\n\nChange-Id: I2fca9d4f6012cc9e4a3c38b169d059cde18fceab\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0207c86622f4f767f6ef9fdb71b1827b7e4bff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/e0207c86622f4f767f6ef9fdb71b1827b7e4bff1"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"ours":"84d690337be11bc46a0a79bb7fd1b1d85f0df88f","theirs":"2bc2b6c61b5463794c8cf545fb081a1b34d4ea90","contains_conflicts":false}},"8fca946ae688994d570148c335bc08d91b2facb1":{"kind":"REWORK","_number":5,"created":"2026-04-16 06:49:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/984804/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/04/984804/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/04/984804/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/04/984804/5"}}},"commit":{"parents":[{"commit":"2bc2b6c61b5463794c8cf545fb081a1b34d4ea90","subject":"Follow up for I170bdb3ffc89bc3307a08da7cd4c7f2793a4e491","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2bc2b6c61b5463794c8cf545fb081a1b34d4ea90"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-15 15:24:45.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-04-16 06:49:24.000000000","tz":0},"subject":"Install pip-check-updates only in jobs requiring that","message":"Install pip-check-updates only in jobs requiring that\n\nChange-Id: I2fca9d4f6012cc9e4a3c38b169d059cde18fceab\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8fca946ae688994d570148c335bc08d91b2facb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8fca946ae688994d570148c335bc08d91b2facb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
