)]}'
{"id":"openstack%2Fproject-config~987373","triplet_id":"openstack%2Fproject-config~master~If0df027d0e880cd6011d3ceb5454871db8bd7531","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2026-05-07 13:41:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0df027d0e880cd6011d3ceb5454871db8bd7531","subject":"Add gerrit and zuul plugin repos for hermes-agent","status":"MERGED","created":"2026-05-05 15:11:09.000000000","updated":"2026-05-07 14:01:03.000000000","submitted":"2026-05-07 13:41:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"987373","meta_rev_id":"7722650b8d4e69f34050382a09688a04df77270a","_number":987373,"virtual_id_number":987373,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-07 13:41:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2026-05-07 13:25:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":2,"date":"2026-05-06 15:38:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2026-05-07 13:25:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-05 15:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-06 15:38:17.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-05-07 13:25:47.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"}],"messages":[{"id":"2e3dba06dbdf5003566e078f3983b95857b427b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-05-05 15:11:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3656318f70a019cd75374842229e44dc7fee2086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 15:16:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22b1601308f34ed99b74da2c85d2e1af\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/19691d2e0c704126a25047ab437d07af : SUCCESS in 4m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/3e742c326cda426a985ea510b2408a25 : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":1},{"id":"03ede15c3eb8390bceaad4f76d9ce7ad9d70423c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-05-05 16:06:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"468ac3d43aa9f50b45e9bfa6c5af8a65456f7175","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:14:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38b952e43d5b41e28e02b80bb9fc910d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/23ed547af21d4e4b9a28c589a25859da : SUCCESS in 4m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dbfc2ab5655d4221a141f6bb6f1e1793 : FAILURE in 3m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1e65f3ac5a6f5c87ef7dcd214fe987cd475ec544","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-05-05 16:47:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"7e6bf9c3a9fe28a2368238857d1d70f575dd133d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-05 16:51:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36d76743667440eaac9a5118d22ce2e5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bd54c36f0ef44924a4590dda3781c8c2 : SUCCESS in 3m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2c768666d7ad4af4abb498dcfa7b5222 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":3},{"id":"28f4069d700916146083b24a95613723275b4851","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-05-06 15:30:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"228322d2ab192c2a3560fceab7237d4b415cf409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 15:34:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/714edd57b2b54e788d2a4ea8dafb0c0a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f793f84edca74965a721ceee21e67cb7 : SUCCESS in 2m 50s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/51701fdbca2740d8869ea8736c4249fe : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":4},{"id":"e18fb3f56127c190feecce10c7c9a87e77558650","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-05-06 15:38:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b550bfb51174658b4e9d45b0a41ffb2d52a436d7","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2026-05-07 13:25:47.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c7f6c423dbb0881416d7600641e43f5639eca8dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 13:30:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"48d424214476cb3e41aee65c10f31fd46919ac1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 13:41:30.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c14522d0207b43249673543ce4cb4341\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/22c1749ebc414023900b92052d96d9a9 : SUCCESS in 4m 39s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8fe716171fe445b48d4c269ebf826707 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":4},{"id":"384e7c1f3a6b3fa90ae7d7aaae6f785453cb9c70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 13:41:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7722650b8d4e69f34050382a09688a04df77270a","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-07 14:01:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8d07615e70f4b78981969936a5b961f\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/0739bced61194cb5bb11a08aa63900ee : SUCCESS in 13m 41s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/6a289a2b7dbd45ccab86bee8966d9089 : SUCCESS in 4m 51s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/2170c0685df64b468419021d5a7f8fff : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"17dd16ef618da633aec7751475b1dd0afaa0a675","revisions":{"eac2437a85a670429c672c3983410ab627c9ebb0":{"kind":"REWORK","_number":1,"created":"2026-05-05 15:11:09.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/73/987373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/987373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/987373/1"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-05 15:08:48.000000000","tz":-420},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-05 15:08:48.000000000","tz":-420},"subject":"Add gerrit and zuul plugin repos for hermes-agent","message":"Add gerrit and zuul plugin repos for hermes-agent\n\nHermes agent is similar to OpenClaw, except:\n\n- It\u0027s in Python\n- It has a backend async learning loop, so it writes skills for\n  itself and integrates into different backend memory providers\n  to self-improve.\n\nMaybe it\u0027s the python thing, but it also feels more solid and less\nyolo.\n\nI\u0027ve already moved a few agents to it and am very pleased with how\nthat went, but I need new versions of my gerrit and zuul interfaces\nsince, you know, this is in python.\n\nChange-Id: If0df027d0e880cd6011d3ceb5454871db8bd7531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eac2437a85a670429c672c3983410ab627c9ebb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eac2437a85a670429c672c3983410ab627c9ebb0"}]},"branch":"refs/heads/master"},"70bd857dff3892838618346dd2e95bfe63d75f9a":{"kind":"REWORK","_number":2,"created":"2026-05-05 16:06:54.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/73/987373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/987373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/987373/2"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-05 15:08:48.000000000","tz":-420},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-05 16:06:50.000000000","tz":-420},"subject":"Add gerrit and zuul plugin repos for hermes-agent","message":"Add gerrit and zuul plugin repos for hermes-agent\n\nHermes agent is similar to OpenClaw, except:\n\n- It\u0027s in Python\n- It has a backend async learning loop, so it writes skills for\n  itself and integrates into different backend memory providers\n  to self-improve.\n\nMaybe it\u0027s the python thing, but it also feels more solid and less\nyolo.\n\nI\u0027ve already moved a few agents to it and am very pleased with how\nthat went, but I need new versions of my gerrit and zuul interfaces\nsince, you know, this is in python.\n\nChange-Id: If0df027d0e880cd6011d3ceb5454871db8bd7531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70bd857dff3892838618346dd2e95bfe63d75f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/70bd857dff3892838618346dd2e95bfe63d75f9a"}]},"branch":"refs/heads/master"},"ccf8698b42d92697ea4f689b5bc38df309be2923":{"kind":"REWORK","_number":3,"created":"2026-05-05 16:47:04.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/73/987373/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/987373/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/987373/3"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-05 15:08:48.000000000","tz":-420},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-05 16:46:58.000000000","tz":-420},"subject":"Add gerrit and zuul plugin repos for hermes-agent","message":"Add gerrit and zuul plugin repos for hermes-agent\n\nHermes agent is similar to OpenClaw, except:\n\n- It\u0027s in Python\n- It has a backend async learning loop, so it writes skills for\n  itself and integrates into different backend memory providers\n  to self-improve.\n\nMaybe it\u0027s the python thing, but it also feels more solid and less\nyolo.\n\nI\u0027ve already moved a few agents to it and am very pleased with how\nthat went, but I need new versions of my gerrit and zuul interfaces\nsince, you know, this is in python.\n\nChange-Id: If0df027d0e880cd6011d3ceb5454871db8bd7531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccf8698b42d92697ea4f689b5bc38df309be2923"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccf8698b42d92697ea4f689b5bc38df309be2923"}]},"branch":"refs/heads/master"},"17dd16ef618da633aec7751475b1dd0afaa0a675":{"kind":"REWORK","_number":4,"created":"2026-05-06 15:30:20.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/73/987373/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/73/987373/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/73/987373/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/73/987373/4"}}},"commit":{"parents":[{"commit":"599675e6f93cd76ee545e13cdcf5b4ef22903a0d","subject":"Merge \"Add translation-jobs to mistral-dashboard\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/599675e6f93cd76ee545e13cdcf5b4ef22903a0d"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-05 15:08:48.000000000","tz":-420},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-06 15:29:33.000000000","tz":-420},"subject":"Add gerrit and zuul plugin repos for hermes-agent","message":"Add gerrit and zuul plugin repos for hermes-agent\n\nHermes agent is similar to OpenClaw, except:\n\n- It\u0027s in Python\n- It has a backend async learning loop, so it writes skills for\n  itself and integrates into different backend memory providers\n  to self-improve.\n\nMaybe it\u0027s the python thing, but it also feels more solid and less\nyolo.\n\nI\u0027ve already moved a few agents to it and am very pleased with how\nthat went, but I need new versions of my gerrit and zuul interfaces\nsince, you know, this is in python.\n\nWhile we\u0027re at it, add hermes-agent and openclaw both to the\ngithub untrusted projects so we can set up cross-repo integration\ntest jobs.\n\nChange-Id: If0df027d0e880cd6011d3ceb5454871db8bd7531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/17dd16ef618da633aec7751475b1dd0afaa0a675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/17dd16ef618da633aec7751475b1dd0afaa0a675"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
