)]}'
{"id":"openstack%2Fproject-config~988655","triplet_id":"openstack%2Fproject-config~master~I50a171a1fdcf3693f24e13d62b9b9e9b6a7bfa0c","project":"openstack/project-config","branch":"master","topic":"retiring-virtualpdu","attention_set":{},"removed_from_attention_set":{"36770":{"account":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"last_update":"2026-05-19 15:47:50.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-05-14 18:12:26.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I50a171a1fdcf3693f24e13d62b9b9e9b6a7bfa0c","subject":"[EOL]: Remove VirtualPDU repository references","status":"MERGED","created":"2026-05-14 17:21:51.000000000","updated":"2026-05-19 16:08:45.000000000","submitted":"2026-05-19 15:47:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"988655-retiring-virtualpdu","meta_rev_id":"9dd32b0d82bbbe301c0804b32d3d68f39be750bd","_number":988655,"virtual_id_number":988655,"owner":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-19 15:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-05-14 18:50:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-05-19 15:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-05-14 18:41:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2026-05-19 14:43:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-14 18:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2026-05-19 15:26:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-14 17:23:24.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2026-05-14 17:30:25.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-05-14 17:39:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-14 18:50:34.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2026-05-19 14:43:39.000000000","updated_by":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"reviewer":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"state":"REVIEWER"},{"updated":"2026-05-19 15:26:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c813ea182317ba7355de70eb8e27f0dcc3a2be8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2026-05-14 17:21:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3165764c5b8cfd80a862aab251c7876786fc10b","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-05-14 17:23:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c0a7f42d2107b34431e535d32d3973e7810d5ed","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-05-14 17:24:05.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"018c184216cbc416d2fb6684a49bd571094b12b9","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-05-14 17:24:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"864c0bb284ddd26fcda50604dd5d6aeb392bdf4c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-05-14 17:30:25.000000000","message":"Patch Set 1: Code-Review-1\n\nTypically this would Depends-On the change in the virtualpdu repo which removes its content and replaces the README file.","accounts_in_message":[],"_revision_number":1},{"id":"1049de90879b0a320ffa1cb5e12c961ebc1165ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:39:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1416c603f8d40878d6c9a151fc5f137\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ee86ad1cda3d4a8d8c8127b636a48473 : SUCCESS in 7m 12s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5199fa3683c74e199d89df5adbf65730 : SUCCESS in 4m 26s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/15846f9d42da4f119af02d0a6f7fcbf0 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"450bb8a5d08b6ad52a55c88d393aa8c33a1d6b8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2026-05-14 17:42:28.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"26bc8999998e51c94d634a070a3d1610b4dcde8c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-05-14 17:44:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"60189c57be010f71e01151bccb360336a1097ba8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-05-14 17:44:26.000000000","message":"Patch Set 2:\n\nOnce the Depends-On change has merged, we\u0027re safe to approve this one.","accounts_in_message":[],"_revision_number":2},{"id":"343775917404393f1ef42e3bc72999010b322f4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2026-05-14 17:44:51.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"820c4fa8e9fd4f09cd97b83ac5c3c357e522e4a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-14 17:53:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce7b88f1a28495281056525f170f53a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a24d077b2e254671a2aef7d45eae8ed9 : SUCCESS in 2m 41s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/04db7bf8c7494071a0b149cea0219418 : SUCCESS in 2m 23s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/990f5e1a0a0841d08e75d78b105f008d : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3},{"id":"4f53c216bab325faad8ae23b88c86f69d4c9f8ed","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-05-14 18:12:26.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"078ac84ed4e60b2272f04691da75925e3923ebef","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2026-05-14 18:41:34.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"62bcc41da42f1f9b178d80796b1ace8ab4768825","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-05-14 18:50:34.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"ba90a5eb8c86d2f0cd411c2cb36d1d67774bb40d","author":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"},"date":"2026-05-19 14:43:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"8c1c03777f2c6e673114ba75286bcc6d0864846f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-05-19 15:26:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3ce77bfb83a53780ce997debe151e0ef59e9631f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 15:28:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bc8af9ae0873e79976d4dfae3e75b8ca490235de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 15:47:50.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50cf4715483c4ad7ae5c7b7116df6314\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/87d5a47e78204f04a7cbbcde5dab591f : SUCCESS in 6m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/b8414e3c9e644de3adc78d870b173d5c : SUCCESS in 4m 11s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/811645d9d55a44878666d57b5c7667e7 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":3},{"id":"561b7e3338e8bb51127989fede06d49197f411e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 15:47:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9dd32b0d82bbbe301c0804b32d3d68f39be750bd","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-19 16:08:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8bc37016c7d4162aed4428dc67ef83c\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/d05289e92324407fb907409a1a6f8d21 : SUCCESS in 14m 18s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/02a1751ce0ef455d8ae0f795608dac28 : SUCCESS in 5m 31s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/57e597583c244601b96b6b66a9686595 : SUCCESS in 5m 57s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/5e558e6438ba4b619550e6be597f1ae3 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8165b29935dae872126ab8a490d124a81cc1381a","revisions":{"114c5886653af70d0a8f3174442d963af96828b0":{"kind":"REWORK","_number":1,"created":"2026-05-14 17:21:51.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/55/988655/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/988655/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/988655/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/988655/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/988655/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/988655/1"}}},"commit":{"parents":[{"commit":"ac4c3f7d635b66fcfbd285c2a694e1e425185365","subject":"Ends project gating for `virtualpdu`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac4c3f7d635b66fcfbd285c2a694e1e425185365"}]}],"author":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2026-05-14 17:16:22.000000000","tz":60},"committer":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2026-05-14 17:16:22.000000000","tz":60},"subject":"Remove VirtualPDU repository references","message":"Remove VirtualPDU repository references\n\nAs part of retiring VirtualPDU, this commit removes the VirtualPDU\nrepository references from the gerrit and zuul configuration files.\n\nChange-Id: I50a171a1fdcf3693f24e13d62b9b9e9b6a7bfa0c\nSigned-off-by: Afonne-CID \u003cafonnepaulc@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/114c5886653af70d0a8f3174442d963af96828b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/114c5886653af70d0a8f3174442d963af96828b0"}]},"branch":"refs/heads/master"},"dda41fd2df76e2e74c69a194e055dc167d973f33":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-05-14 17:42:28.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/55/988655/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/988655/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/988655/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/988655/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/988655/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/988655/2"}}},"commit":{"parents":[{"commit":"ac4c3f7d635b66fcfbd285c2a694e1e425185365","subject":"Ends project gating for `virtualpdu`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac4c3f7d635b66fcfbd285c2a694e1e425185365"}]}],"author":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2026-05-14 17:16:22.000000000","tz":60},"committer":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2026-05-14 17:41:50.000000000","tz":60},"subject":"Remove VirtualPDU repository references","message":"Remove VirtualPDU repository references\n\nAs part of retiring VirtualPDU, this commit removes the VirtualPDU\nrepository references from the gerrit and zuul configuration files.\n\nDepends-On: https://review.opendev.org/c/openstack/virtualpdu/+/984162\nChange-Id: I50a171a1fdcf3693f24e13d62b9b9e9b6a7bfa0c\nSigned-off-by: Afonne-CID \u003cafonnepaulc@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dda41fd2df76e2e74c69a194e055dc167d973f33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/dda41fd2df76e2e74c69a194e055dc167d973f33"}]},"branch":"refs/heads/master"},"8165b29935dae872126ab8a490d124a81cc1381a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-05-14 17:44:51.000000000","uploader":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"ref":"refs/changes/55/988655/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/55/988655/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/988655/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/988655/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/55/988655/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/55/988655/3"}}},"commit":{"parents":[{"commit":"ac4c3f7d635b66fcfbd285c2a694e1e425185365","subject":"Ends project gating for `virtualpdu`","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ac4c3f7d635b66fcfbd285c2a694e1e425185365"}]}],"author":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2026-05-14 17:16:22.000000000","tz":60},"committer":{"name":"Afonne-CID","email":"afonnepaulc@gmail.com","date":"2026-05-14 17:44:10.000000000","tz":60},"subject":"[EOL]: Remove VirtualPDU repository references","message":"[EOL]: Remove VirtualPDU repository references\n\nAs part of retiring VirtualPDU, this commit removes the VirtualPDU\nrepository references from the gerrit and zuul configuration files.\n\nDepends-On: https://review.opendev.org/c/openstack/virtualpdu/+/984162\nChange-Id: I50a171a1fdcf3693f24e13d62b9b9e9b6a7bfa0c\nSigned-off-by: Afonne-CID \u003cafonnepaulc@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8165b29935dae872126ab8a490d124a81cc1381a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8165b29935dae872126ab8a490d124a81cc1381a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35929,"name":"Mahnoor Asghar","display_name":"Mahnoor Asghar","email":"masghar@redhat.com","username":"mahnoorasghar"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
