)]}'
{"id":"openstack%2Fproject-config~989417","triplet_id":"openstack%2Fproject-config~master~Ia5e3f94840e6fad594ee3ad00e9f41b438ab3561","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2026-05-28 18:06:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia5e3f94840e6fad594ee3ad00e9f41b438ab3561","subject":"Import drizzle revival and add a bindep-rs support project","status":"MERGED","created":"2026-05-20 15:40:31.000000000","updated":"2026-05-28 18:18:50.000000000","submitted":"2026-05-28 18:06:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989417","meta_rev_id":"de2704dce5c8f99fa009466c3de3851238051957","_number":989417,"virtual_id_number":989417,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-28 18:06:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2026-05-28 17:41:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-28 17:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":1,"date":"2026-05-28 17:53:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-20 15:50:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-28 17:04:06.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-05-28 17:41:36.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"}],"messages":[{"id":"65bc867c4d6a3c70e5bc6274a325661422a27ba7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-05-20 15:40:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a94dc388ad114c3286d59232022726a283656ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 15:50:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f668eaad1944e80aed80fa8f91d4e2c\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8b2930205b4144e6a8da1c076316fd53 : SUCCESS in 4m 27s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0c52a864faa34f85bd708df9bbdec0df : FAILURE in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c25058ffc869adfbb3e4443c7a21ac31c342977f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-05-20 15:57:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7873ccf93e9cb8c8ff41b5dc03e5f0485a3e2e9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 16:03:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c580182925594d70acb367cc9f63e77a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/64da133a72364840be1fd1143fc7027a : SUCCESS in 2m 53s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e517b4de3cce446eb462e0ac2a52e7ef : FAILURE in 2m 18s","accounts_in_message":[],"_revision_number":2},{"id":"ac1906c4a827d8ce2848013217a033caa99af156","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-05-20 18:01:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5d77514b1543772bb29f1971f1d1f86a47c4aaf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-20 18:10:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb758be9ef54492b9119c7b21a9e0c15\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/22a39afe64f545c594d80ad92aedaf65 : SUCCESS in 4m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fbb9669482ed4b9a88b15b628f76cc7a : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":3},{"id":"93f9ebb303040b0bce938f2c15ab3e1997ba812d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-05-25 19:40:49.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"618ff0e9e4c5592dc0baf56908f41a9928b65a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 19:45:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/040e05d0a0f349e5baae401189a96160\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f28efd92e1dc4417ac26fb8f1349acd9 : SUCCESS in 3m 21s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ec1dfc44a71940a78dfeba0950432045 : FAILURE in 3m 28s","accounts_in_message":[],"_revision_number":4},{"id":"dd4c21c242730e8f652b5547615125a91c090767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-05-25 20:25:21.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"ca30ae56e5cd48c4b8fbdf7da662a77e942acb78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-25 20:33:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45d146f52121429f97e0c17483eb339a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/01435757d9ce4f59b09d1ff8e92dd850 : SUCCESS in 4m 35s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/014a0eca3413403f8e98fc65ffed12c2 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":5},{"id":"b5d6d40242ce5980c86eaf42d07ee688b528eb85","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-05-28 17:04:06.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"2c1d07424f4b9ccf46dfb0b85f0450025d458f98","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2026-05-28 17:41:36.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f5eef27db8dfd6a019984e4f743b5190b9c9e90b","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2026-05-28 17:53:59.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"bc242250fc73d2d273f9ca84e7eb5300156cb859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 17:54:12.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e2f2d7cd064a0872171a97777f7af2636d1ad894","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 18:06:27.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e4ea230802043f08709df41736eaa05\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/47063a34f07c4465a8182697daca7539 : SUCCESS in 3m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f0f454fd238242b8ad4aacdd7fc4c676 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":5},{"id":"4a9ef42a1c0eb259262289bd76c8752daf80c594","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 18:06:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"de2704dce5c8f99fa009466c3de3851238051957","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-28 18:18:50.000000000","message":"Patch Set 5:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/219b0da628d14c829385b3b17a4949a5\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/bc68674af112497892179c15cbd0b981 : SUCCESS in 10m 29s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/e724bf923a284598991cbcd09deb494f : FAILURE in 8m 44s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/13c4a417007a4f1fbd005f8ea465020a : SKIPPED Skipped due to failed job infra-prod-manage-projects","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"300dd61af951c535d383f0bc69b04f0f7313f443","revisions":{"5a72c63714b7ac7b1382f432098960b09739ab8d":{"kind":"REWORK","_number":1,"created":"2026-05-20 15:40:31.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/17/989417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/17/989417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/17/989417/1"}}},"commit":{"parents":[{"commit":"4ceafe4ac7389a5a0b0b03adf70d3083533ed9db","subject":"Merge \"Allow registered users to toggleWipState in OpenStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ceafe4ac7389a5a0b0b03adf70d3083533ed9db"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-20 15:36:26.000000000","tz":-420},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-20 15:36:26.000000000","tz":-420},"subject":"Import drizzle revival and add a bindep-rs support project","message":"Import drizzle revival and add a bindep-rs support project\n\nIn the days of yore there lived a project called Drizzle which was\na crazy fork of MySQL funded by the dot in dot com. The dev practices\nof drizzle became the dev practices of openstack, and so is ultimately\na step in the direct ancestry of opendev itself.\n\nSadly, with mindshare shifted to OpenStack, Drizzle withered and was\nlost to time.\n\nBut now that we can throw LLMs at problems for only the cost of\ndestroying the planet, it feels like a time to revive Drizzle and\nbring it into what should be its natural habitat, the project gating\nsystems that actualy encoded what were previously just dev community\nmanual practices.\n\nAlso, yes, I\u0027m nuts. But I got a patch from shrews already!\n\nChange-Id: Ia5e3f94840e6fad594ee3ad00e9f41b438ab3561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a72c63714b7ac7b1382f432098960b09739ab8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5a72c63714b7ac7b1382f432098960b09739ab8d"}]},"branch":"refs/heads/master"},"b95188244034b9881db8a1f47e076202071b38bf":{"kind":"REWORK","_number":2,"created":"2026-05-20 15:57:02.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/17/989417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/17/989417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/17/989417/2"}}},"commit":{"parents":[{"commit":"4ceafe4ac7389a5a0b0b03adf70d3083533ed9db","subject":"Merge \"Allow registered users to toggleWipState in OpenStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ceafe4ac7389a5a0b0b03adf70d3083533ed9db"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-20 15:36:26.000000000","tz":-420},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-20 15:56:58.000000000","tz":-420},"subject":"Import drizzle revival and add a bindep-rs support project","message":"Import drizzle revival and add a bindep-rs support project\n\nIn the days of yore there lived a project called Drizzle which was\na crazy fork of MySQL funded by the dot in dot com. The dev practices\nof drizzle became the dev practices of openstack, and so is ultimately\na step in the direct ancestry of opendev itself.\n\nSadly, with mindshare shifted to OpenStack, Drizzle withered and was\nlost to time.\n\nBut now that we can throw LLMs at problems for only the cost of\ndestroying the planet, it feels like a time to revive Drizzle and\nbring it into what should be its natural habitat, the project gating\nsystems that actualy encoded what were previously just dev community\nmanual practices.\n\nAlso, yes, I\u0027m nuts. But I got a patch from shrews already!\n\nChange-Id: Ia5e3f94840e6fad594ee3ad00e9f41b438ab3561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b95188244034b9881db8a1f47e076202071b38bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b95188244034b9881db8a1f47e076202071b38bf"}]},"branch":"refs/heads/master"},"8770138331111cb76824cc07c78aa05fd26fa25d":{"kind":"REWORK","_number":3,"created":"2026-05-20 18:01:07.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/17/989417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/17/989417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/17/989417/3"}}},"commit":{"parents":[{"commit":"4ceafe4ac7389a5a0b0b03adf70d3083533ed9db","subject":"Merge \"Allow registered users to toggleWipState in OpenStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ceafe4ac7389a5a0b0b03adf70d3083533ed9db"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-20 15:36:26.000000000","tz":-420},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-20 18:01:01.000000000","tz":-420},"subject":"Import drizzle revival and add a bindep-rs support project","message":"Import drizzle revival and add a bindep-rs support project\n\nIn the days of yore there lived a project called Drizzle which was\na crazy fork of MySQL funded by the dot in dot com. The dev practices\nof drizzle became the dev practices of openstack, and so is ultimately\na step in the direct ancestry of opendev itself.\n\nSadly, with mindshare shifted to OpenStack, Drizzle withered and was\nlost to time.\n\nBut now that we can throw LLMs at problems for only the cost of\ndestroying the planet, it feels like a time to revive Drizzle and\nbring it into what should be its natural habitat, the project gating\nsystems that actualy encoded what were previously just dev community\nmanual practices.\n\nAlso, yes, I\u0027m nuts. But I got a patch from shrews already!\n\nChange-Id: Ia5e3f94840e6fad594ee3ad00e9f41b438ab3561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8770138331111cb76824cc07c78aa05fd26fa25d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8770138331111cb76824cc07c78aa05fd26fa25d"}]},"branch":"refs/heads/master"},"2526d56294129e972a968e81372c69ec3c94aa5f":{"kind":"REWORK","_number":4,"created":"2026-05-25 19:40:49.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/17/989417/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/17/989417/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/17/989417/4"}}},"commit":{"parents":[{"commit":"4ceafe4ac7389a5a0b0b03adf70d3083533ed9db","subject":"Merge \"Allow registered users to toggleWipState in OpenStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ceafe4ac7389a5a0b0b03adf70d3083533ed9db"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-20 15:36:26.000000000","tz":-420},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-25 19:31:59.000000000","tz":-420},"subject":"Import drizzle revival and add a bindep-rs support project","message":"Import drizzle revival and add a bindep-rs support project\n\nIn the days of yore there lived a project called Drizzle which was\na crazy fork of MySQL funded by the dot in dot com. The dev practices\nof drizzle became the dev practices of openstack, and so is ultimately\na step in the direct ancestry of opendev itself.\n\nSadly, with mindshare shifted to OpenStack, Drizzle withered and was\nlost to time.\n\nBut now that we can throw LLMs at problems for only the cost of\ndestroying the planet, it feels like a time to revive Drizzle and\nbring it into what should be its natural habitat, the project gating\nsystems that actualy encoded what were previously just dev community\nmanual practices.\n\nAlso, yes, I\u0027m nuts. But I got a patch from shrews already!\n\nAlso, yes, that\u0027s a perl module. The drizzle test runner is written\nin perl and uses it. What\u0027s old is new again, right?\n\nChange-Id: Ia5e3f94840e6fad594ee3ad00e9f41b438ab3561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2526d56294129e972a968e81372c69ec3c94aa5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2526d56294129e972a968e81372c69ec3c94aa5f"}]},"branch":"refs/heads/master"},"300dd61af951c535d383f0bc69b04f0f7313f443":{"kind":"REWORK","_number":5,"created":"2026-05-25 20:25:21.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/17/989417/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/17/989417/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/17/989417/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/17/989417/5"}}},"commit":{"parents":[{"commit":"4ceafe4ac7389a5a0b0b03adf70d3083533ed9db","subject":"Merge \"Allow registered users to toggleWipState in OpenStack\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4ceafe4ac7389a5a0b0b03adf70d3083533ed9db"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-20 15:36:26.000000000","tz":-420},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-05-25 20:25:14.000000000","tz":-420},"subject":"Import drizzle revival and add a bindep-rs support project","message":"Import drizzle revival and add a bindep-rs support project\n\nIn the days of yore there lived a project called Drizzle which was\na crazy fork of MySQL funded by the dot in dot com. The dev practices\nof drizzle became the dev practices of openstack, and so is ultimately\na step in the direct ancestry of opendev itself.\n\nSadly, with mindshare shifted to OpenStack, Drizzle withered and was\nlost to time.\n\nBut now that we can throw LLMs at problems for only the cost of\ndestroying the planet, it feels like a time to revive Drizzle and\nbring it into what should be its natural habitat, the project gating\nsystems that actualy encoded what were previously just dev community\nmanual practices.\n\nAlso, yes, I\u0027m nuts. But I got a patch from shrews already!\n\nAlso, yes, that\u0027s a perl module. The drizzle test runner is written\nin perl and uses it. What\u0027s old is new again, right?\n\nChange-Id: Ia5e3f94840e6fad594ee3ad00e9f41b438ab3561\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/300dd61af951c535d383f0bc69b04f0f7313f443"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/300dd61af951c535d383f0bc69b04f0f7313f443"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
