)]}'
{"id":"openstack%2Fproject-config~990078","triplet_id":"openstack%2Fproject-config~master~I8c8ecb3bf34fb57808ec531a40a258db196585e0","project":"openstack/project-config","branch":"master","topic":"remove-pqc-gerritbot-projects","attention_set":{},"removed_from_attention_set":{"35125":{"account":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"last_update":"2026-05-26 13:08:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8c8ecb3bf34fb57808ec531a40a258db196585e0","subject":"Remove gerritbot project notifications from #openstack-pqc","status":"MERGED","created":"2026-05-26 09:59:33.000000000","updated":"2026-05-26 13:13:36.000000000","submitted":"2026-05-26 13:08:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990078-remove-pqc-gerritbot-projects","meta_rev_id":"dc1399dd3c09cfd34d7059d517fa3bf1e39abb80","_number":990078,"virtual_id_number":990078,"owner":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-26 13:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-26 12:57:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-26 12:57:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-26 10:11:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-26 12:57:44.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"24a9d70e9925d4380f24f515a1acf17a7f6b8ba7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2026-05-26 09:59:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f046076feb5144b959df7369b28b6b7e1f7d36b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 10:11:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecac00b6f6874c03907ef23cd5c2e33f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1cab9e0cb6004cbd9aba032c6a5397fa : SUCCESS in 5m 44s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/c5c3c21f667048fbb76911e3f9a7684d : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6632d41b4a539a8a333510ff3caa87e34228799a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-05-26 12:57:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e3fee0511355332153cf9f906a8575538eb93d84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 12:57:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73dd73e0eee63f0ab604858de7dfa6a84df2aa96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 13:08:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deec0b99178143c4a06590b5db759ea4\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ce3bda2902444d988cd7ab6f95b3b500 : SUCCESS in 3m 22s\n- project-config-irc-access https://zuul.opendev.org/t/openstack/build/2facffebbbd3496baed6666fbdf351ec : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"defc2952500a910dd0c7763cdb3cfa910d7bf089","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 13:08:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dc1399dd3c09cfd34d7059d517fa3bf1e39abb80","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-26 13:13:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/299316de7164410c89b5f819b365e327\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/8b813e5b53d84f4fb6caa762317d0499 : SUCCESS in 4m 32s\n- infra-prod-service-eavesdrop https://zuul.opendev.org/t/openstack/build/6e170deef1754fbcb1cab320e50ff4c7 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"93a08246c24bea6015348eee412efcbf1df9aeed","revisions":{"93a08246c24bea6015348eee412efcbf1df9aeed":{"kind":"REWORK","_number":1,"created":"2026-05-26 09:59:33.000000000","uploader":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"ref":"refs/changes/78/990078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/78/990078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/990078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/990078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/78/990078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/78/990078/1"}}},"commit":{"parents":[{"commit":"c1d17d3831b1beeb1d55e9a41d2a708a7f08f461","subject":"Allow Trove Cores to change Backport-Candidate labels","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1d17d3831b1beeb1d55e9a41d2a708a7f08f461"}]}],"author":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2026-05-26 09:56:08.000000000","tz":0},"committer":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2026-05-26 09:56:08.000000000","tz":0},"subject":"Remove gerritbot project notifications from #openstack-pqc","message":"Remove gerritbot project notifications from #openstack-pqc\n\nThe Gerrit notifications for all monitored projects were flooding\nthe channel with too much noise, making it difficult to hold\nconversations. Remove the project list from gerritbot configuration\nwhile keeping the channel registered in accessbot.\n\nChange-Id: I8c8ecb3bf34fb57808ec531a40a258db196585e0\nSigned-off-by: Mauricio Harley \u003cmharley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/93a08246c24bea6015348eee412efcbf1df9aeed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/93a08246c24bea6015348eee412efcbf1df9aeed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
