)]}'
{"id":"openstack%2Fproject-config~990586","triplet_id":"openstack%2Fproject-config~master~Ieebccb729f94ca82b66f8d55be01948898c48626","project":"openstack/project-config","branch":"master","topic":"project-yaml-normalization","hashtags":[],"change_id":"Ieebccb729f94ca82b66f8d55be01948898c48626","subject":"Normalize projects.yaml","status":"MERGED","created":"2026-05-29 02:25:52.000000000","updated":"2026-05-31 13:05:05.000000000","submitted":"2026-05-31 12:59:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990586-project-yaml-normalization","meta_rev_id":"6b161e0e218917ada32ac5d187bbc9c4662e3c67","_number":990586,"virtual_id_number":990586,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-31 12:59:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-05-31 12:48:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-05-31 12:48:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-29 03:01:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-31 12:48:35.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"1c6b6d4e698c58d30fe2338a1f0674608eef11f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-29 02:25:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98dd5732bb0c198196332812a6e6339d0b6ed612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 03:01:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48e75cec04d640ad9287c85491df7385\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/11e5b08f57394da8962903550afd5eef : SUCCESS in 4m 54s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2c890a88bfba48bab789a662f3b8fdc2 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8d16fb5021fd094ae72b44fbc58f90391f49492a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-30 02:56:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bd634fd0460b7b05ca416b96bb8d2612d00be0f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-30 03:04:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d70f551e53d43aa826cbefea040cd0a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0106b3746ee445b9be831061f0f4e90b : SUCCESS in 4m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/ac028cb6a7b6455d9d1ce79b810a8c33 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":2},{"id":"5f6553fffb2914aa524a53718cdece7ccb831a1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2026-05-31 02:23:37.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"b2f169f53fadef246f747ac969f0813c3349f23d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 02:54:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c33c89c8a5104f07a22c90971dc1e495\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/fe8a5b8cfb794cd2a876d2465aae8095 : SUCCESS in 3m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/8e42da6fc49947899ddd03ef58c31b11 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":3},{"id":"5561d924a7df757cc8b799fd27de367e2feaff78","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-05-31 12:48:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"5158a46ec3da6253dec5f9c6e248cdba291887db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 12:48:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a9dedcdca06c7185958720656bfa7b0b208bf9fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 12:59:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42fbfe00a7074859ac20c1fe9edfa0d5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/0be05bcf4c2043f79622419226c82417 : SUCCESS in 4m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dadb94ef5c0a42bf9e40e1fdff9da7d0 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"769408aeb9f7b56c7435fca441f2103cd03c5eb5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 12:59:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6b161e0e218917ada32ac5d187bbc9c4662e3c67","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-31 13:05:05.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad34afe88ecb41909f517bb2763fe56b\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/c7fdd04a47554c7493eca0d36826a5c7 : SUCCESS in 5m 47s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/41e1f051c5f149e680e72d9bcfbb854a : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ee488c06df4306211d2180c23ca8e8badbb98e4a","revisions":{"3912e75ec0d52dee8d5064507c75c3ac1f9e8f76":{"kind":"REWORK","_number":1,"created":"2026-05-29 02:25:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/990586/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/86/990586/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/990586/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/990586/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/990586/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/86/990586/1"}}},"commit":{"parents":[{"commit":"4af965d91b99b59300517ccf6015aa22a2769a6b","subject":"Merge \"Import drizzle revival and add a bindep-rs support project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4af965d91b99b59300517ccf6015aa22a2769a6b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-29 02:25:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-29 02:25:50.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: Ieebccb729f94ca82b66f8d55be01948898c48626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3912e75ec0d52dee8d5064507c75c3ac1f9e8f76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3912e75ec0d52dee8d5064507c75c3ac1f9e8f76"}]},"branch":"refs/heads/master"},"a62792b61b01495e7e6a38a9613d50e1892f1468":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-05-30 02:56:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/990586/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/86/990586/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/990586/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/990586/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/990586/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/86/990586/2"}}},"commit":{"parents":[{"commit":"7bc612054525f06d8f01ef7aa1e7eb0842dc41bf","subject":"Merge \"Add starlingx/app-machine-operator project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bc612054525f06d8f01ef7aa1e7eb0842dc41bf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-30 02:56:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-30 02:56:15.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: Ieebccb729f94ca82b66f8d55be01948898c48626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a62792b61b01495e7e6a38a9613d50e1892f1468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a62792b61b01495e7e6a38a9613d50e1892f1468"}]},"branch":"refs/heads/master"},"ee488c06df4306211d2180c23ca8e8badbb98e4a":{"kind":"NO_CHANGE","_number":3,"created":"2026-05-31 02:23:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/86/990586/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/86/990586/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/990586/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/990586/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/86/990586/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/86/990586/3"}}},"commit":{"parents":[{"commit":"7bc612054525f06d8f01ef7aa1e7eb0842dc41bf","subject":"Merge \"Add starlingx/app-machine-operator project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7bc612054525f06d8f01ef7aa1e7eb0842dc41bf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-31 02:23:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2026-05-31 02:23:35.000000000","tz":0},"subject":"Normalize projects.yaml","message":"Normalize projects.yaml\n\nChange-Id: Ieebccb729f94ca82b66f8d55be01948898c48626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee488c06df4306211d2180c23ca8e8badbb98e4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ee488c06df4306211d2180c23ca8e8badbb98e4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
