)]}'
{"id":"openstack%2Fproject-config~990651","triplet_id":"openstack%2Fproject-config~master~Ifd3f47ac2bbf88ca9f75a763026dec320338f57f","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"28676":{"account":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"last_update":"2026-06-02 16:13:15.000000000","reason":"\u003cGERRIT_ACCOUNT_28676\u003e replied on the change","reason_account":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"}},"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2026-06-02 16:27:07.000000000","reason":"\u003cGERRIT_ACCOUNT_28410\u003e replied on the change","reason_account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}},"31292":{"account":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"last_update":"2026-06-02 16:44:08.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-05-29 14:57:33.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ifd3f47ac2bbf88ca9f75a763026dec320338f57f","subject":"Add repo app-ejbca for starlingx","status":"MERGED","created":"2026-05-29 13:43:15.000000000","updated":"2026-06-02 17:16:09.000000000","submitted":"2026-06-02 16:44:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"990651-new-project","meta_rev_id":"fc6a46aca472a9774116868e27a6abfb7b40731a","_number":990651,"virtual_id_number":990651,"owner":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-02 16:44:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"all":[{"value":1,"date":"2026-06-02 16:13:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},{"value":1,"date":"2026-06-02 16:27:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":1,"date":"2026-06-01 16:21:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-02 15:41:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},{"value":2,"date":"2026-05-29 14:57:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},{"value":1,"date":"2026-06-02 16:35:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},{"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-29 13:45:09.000000000","updated_by":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"reviewer":{"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},"state":"CC"},{"updated":"2026-05-29 13:45:09.000000000","updated_by":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"reviewer":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"state":"CC"},{"updated":"2026-05-29 13:45:09.000000000","updated_by":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"reviewer":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"state":"CC"},{"updated":"2026-05-29 13:45:09.000000000","updated_by":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"},{"updated":"2026-05-29 13:45:09.000000000","updated_by":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-05-29 13:50:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-01 16:21:00.000000000","updated_by":{"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},"reviewer":{"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},"state":"REVIEWER"},{"updated":"2026-06-02 15:41:24.000000000","updated_by":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"reviewer":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"state":"REVIEWER"},{"updated":"2026-06-02 15:42:43.000000000","updated_by":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"reviewer":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"state":"REVIEWER"}],"messages":[{"id":"76aa8e79a88bd8e83fea2e65e050d296aa4c1314","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"date":"2026-05-29 13:43:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e76a08609d4c3be278b4548d2e8361aa591c1f12","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"date":"2026-05-29 13:44:11.000000000","message":"Topic set to new-project","accounts_in_message":[],"_revision_number":1},{"id":"8abb18aa9fec524d292b8141d374c25e9de1ddcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 13:50:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1619034d71d4ed7847f79a6e39c8a52\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/9c872d4289bb4cc8840278f2284d8d3a : SUCCESS in 5m 15s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/72e859a9530b4638bff9adf5e60a6d9e : FAILURE in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a2f7f16fd9a06dc36174bec0b81ca927df43dd06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"date":"2026-05-29 13:53:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"587826bb02d12951dc08ce1ad1feb5b903dca12e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 14:03:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cee1d44590f4032926ae8d8840b471e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e98526a54295442ba43d58b9a96480db : SUCCESS in 4m 55s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/12bf42e5ce034804a7f0dd35367c19e2 : FAILURE in 3m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b75e34add80aaf0d51a5356f28107511f49fa203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"date":"2026-05-29 14:06:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"5d40e471ea82bceb0b95ffa9799dee88f1f7f8e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-29 14:17:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4917cbeb6aa4027b672607492389dde\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/028f2c61d01446beb9f43fb774e85054 : SUCCESS in 3m 36s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/1e9809578ec34d6592bd1a2df5571bb7 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"f251d2306400dc16994a6e9ccb93a9f7103d66cb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-05-29 14:57:33.000000000","message":"Patch Set 3: Code-Review+2\n\nWe can approve this request once Scott acknowledges.","accounts_in_message":[],"_revision_number":3},{"id":"d93beb41ad0d588a7a09aa5cd70f87eb83214892","author":{"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},"date":"2026-06-01 16:21:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"a7d5fe363121c2e50c336b6046711382fbaab790","author":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"date":"2026-06-02 15:41:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"dfe1038c58596ea53ed6208929d7a69b75cdbf30","author":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"date":"2026-06-02 16:13:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"03728f3ee3d7b81a90b67ae7e256dbe208452f51","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2026-06-02 16:27:07.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"6c1a10b4af2aa439cc2215c5184238a8b717336c","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-02 16:35:31.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d9352ebb988eca44b3c5fe5114ebbaaa700458d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 16:35:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0cccc1bd68fb96a9556ed818ab37981d7f231b9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 16:44:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15ad98d392704945aeda305e5fcd6680\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/706732b49b184b37a7b21ec48af64362 : SUCCESS in 2m 45s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/6f2961b85b104a5f89b62a1b9ab2416d : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":3},{"id":"151833828f853f3d7f31b1bce9e160f6dedc4183","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 16:44:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9f9c82be4c571293e899145bf9717a8d85fa4efe","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-02 17:01:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f678b22850f04f5c89ce4dea23354719\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/dd88433093144accaf752c20b12a9e9d : SUCCESS in 13m 41s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/ca58424ee37749f39ae1246f1964929d : SUCCESS in 4m 55s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/ad1e6b8229d1447f94a42ec54369f10e : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":3},{"id":"fc6a46aca472a9774116868e27a6abfb7b40731a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-02 17:16:09.000000000","message":"Patch Set 3:\n\nI\u0027ve added Scott as the initial member of the starlingx-app-ejbca-core group now.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ad88a1014286241951d42b847e68679f50752e64","revisions":{"5aa82b58b261b83eb1208b10cb0d606317f89a79":{"kind":"REWORK","_number":1,"created":"2026-05-29 13:43:15.000000000","uploader":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"ref":"refs/changes/51/990651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/51/990651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/990651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/990651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/990651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/51/990651/1"}}},"commit":{"parents":[{"commit":"4af965d91b99b59300517ccf6015aa22a2769a6b","subject":"Merge \"Import drizzle revival and add a bindep-rs support project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4af965d91b99b59300517ccf6015aa22a2769a6b"}]}],"author":{"name":"amantri","email":"ayyappa.mantri@windriver.com","date":"2026-05-29 13:38:12.000000000","tz":-240},"committer":{"name":"amantri","email":"ayyappa.mantri@windriver.com","date":"2026-05-29 13:41:31.000000000","tz":-240},"subject":"Add repo app-ejbca for starlingx","message":"Add repo app-ejbca for starlingx\n\nThis commit adds a new repo named app-ejbca, an EJBCA PKI system\napplication repo for StarlingX.\n\nStory: 2011712\nTask: 54712\n\nChange-Id: Ifd3f47ac2bbf88ca9f75a763026dec320338f57f\nSigned-off-by: amantri \u003cayyappa.mantri@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5aa82b58b261b83eb1208b10cb0d606317f89a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5aa82b58b261b83eb1208b10cb0d606317f89a79"}]},"branch":"refs/heads/master"},"c2a75910a223fbbbacbb39c47169d9987ea56ff8":{"kind":"REWORK","_number":2,"created":"2026-05-29 13:53:55.000000000","uploader":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"ref":"refs/changes/51/990651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/51/990651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/990651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/990651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/990651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/51/990651/2"}}},"commit":{"parents":[{"commit":"4af965d91b99b59300517ccf6015aa22a2769a6b","subject":"Merge \"Import drizzle revival and add a bindep-rs support project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4af965d91b99b59300517ccf6015aa22a2769a6b"}]}],"author":{"name":"amantri","email":"ayyappa.mantri@windriver.com","date":"2026-05-29 13:38:12.000000000","tz":-240},"committer":{"name":"amantri","email":"ayyappa.mantri@windriver.com","date":"2026-05-29 13:53:53.000000000","tz":-240},"subject":"Add repo app-ejbca for starlingx","message":"Add repo app-ejbca for starlingx\n\nThis commit adds a new repo named app-ejbca, an EJBCA PKI system\napplication repo for StarlingX.\n\nStory: 2011712\nTask: 54712\n\nChange-Id: Ifd3f47ac2bbf88ca9f75a763026dec320338f57f\nSigned-off-by: amantri \u003cayyappa.mantri@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c2a75910a223fbbbacbb39c47169d9987ea56ff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c2a75910a223fbbbacbb39c47169d9987ea56ff8"}]},"branch":"refs/heads/master"},"ad88a1014286241951d42b847e68679f50752e64":{"kind":"REWORK","_number":3,"created":"2026-05-29 14:06:38.000000000","uploader":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"ref":"refs/changes/51/990651/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/51/990651/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/990651/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/990651/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/51/990651/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/51/990651/3"}}},"commit":{"parents":[{"commit":"4af965d91b99b59300517ccf6015aa22a2769a6b","subject":"Merge \"Import drizzle revival and add a bindep-rs support project\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4af965d91b99b59300517ccf6015aa22a2769a6b"}]}],"author":{"name":"amantri","email":"ayyappa.mantri@windriver.com","date":"2026-05-29 13:38:12.000000000","tz":-240},"committer":{"name":"amantri","email":"ayyappa.mantri@windriver.com","date":"2026-05-29 14:06:35.000000000","tz":-240},"subject":"Add repo app-ejbca for starlingx","message":"Add repo app-ejbca for starlingx\n\nThis commit adds a new repo named app-ejbca, an EJBCA PKI system\napplication repo for StarlingX.\n\nStory: 2011712\nTask: 54712\n\nChange-Id: Ifd3f47ac2bbf88ca9f75a763026dec320338f57f\nSigned-off-by: amantri \u003cayyappa.mantri@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad88a1014286241951d42b847e68679f50752e64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ad88a1014286241951d42b847e68679f50752e64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
