)]}'
{"id":"openstack%2Fproject-config~991438","triplet_id":"openstack%2Fproject-config~master~I30a3df6bf47be9f952427c9730cbaa6971b00122","project":"openstack/project-config","branch":"master","topic":"oauth2-proxy-app","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-06-03 20:45:13.000000000","reason":"Change was submitted"},"36727":{"account":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"last_update":"2026-06-03 20:45:13.000000000","reason":"Change was submitted"},"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2026-06-03 20:19:58.000000000","reason":"\u003cGERRIT_ACCOUNT_28410\u003e replied on the change","reason_account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}},"38911":{"account":{"_account_id":38911,"name":"Luiz Andre Ventura Gosling","display_name":"Luiz Andre Ventura Gosling","email":"luizandre.gosling@windriver.com","username":"lgosling"},"last_update":"2026-06-03 20:45:13.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-06-03 19:45:36.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I30a3df6bf47be9f952427c9730cbaa6971b00122","subject":"Add repo oauth2-proxy-app for starlingx","status":"MERGED","created":"2026-06-03 15:38:11.000000000","updated":"2026-06-03 21:00:08.000000000","submitted":"2026-06-03 20:45:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991438-oauth2-proxy-app","meta_rev_id":"8547005a5ea47ee0b17a48986f86820639d015ec","_number":991438,"virtual_id_number":991438,"owner":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-03 20:45:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"all":[{"value":1,"date":"2026-06-03 20:19:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-03 19:45:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-03 20:39:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":38911,"name":"Luiz Andre Ventura Gosling","display_name":"Luiz Andre Ventura Gosling","email":"luizandre.gosling@windriver.com","username":"lgosling"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"_account_id":38911,"name":"Luiz Andre Ventura Gosling","display_name":"Luiz Andre Ventura Gosling","email":"luizandre.gosling@windriver.com","username":"lgosling"}],"CC":[{"_account_id":35727,"name":"Henrique Spindola","display_name":"Henrique Spindola","email":"henrique.spindolacorreia@windriver.com","username":"hspindol"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-03 16:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-03 19:30:02.000000000","updated_by":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"reviewer":{"_account_id":35727,"name":"Henrique Spindola","display_name":"Henrique Spindola","email":"henrique.spindolacorreia@windriver.com","username":"hspindol"},"state":"CC"},{"updated":"2026-06-03 19:30:02.000000000","updated_by":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"reviewer":{"_account_id":38911,"name":"Luiz Andre Ventura Gosling","display_name":"Luiz Andre Ventura Gosling","email":"luizandre.gosling@windriver.com","username":"lgosling"},"state":"REVIEWER"},{"updated":"2026-06-03 19:30:02.000000000","updated_by":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-06-03 19:30:02.000000000","updated_by":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2026-06-03 19:30:02.000000000","updated_by":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"}],"messages":[{"id":"df62d220c4561370e803164db3e0aa5a62576773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"date":"2026-06-03 15:38:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60a7b6a25c764a6dfe25812db3b9b5dde061e436","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"date":"2026-06-03 15:38:36.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"03efd9991d3f1fd9310a6f06efb2ebc2b01292e1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"date":"2026-06-03 15:40:00.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ed72939ac327f29078ed07da001399701ea72c93","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"date":"2026-06-03 15:44:33.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"51e0ddea1e2b15c217471b1c914c47a9a2eb291a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"date":"2026-06-03 15:44:55.000000000","message":"Topic set to oauth2-proxy-app","accounts_in_message":[],"_revision_number":3},{"id":"fe02a87899536ee63267dcdac40e273fe0439368","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 16:20:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5fc5448fb4546a8944ae572cd2a866e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/406bb7ddbb7246d6a3e187cf67885862 : SUCCESS in 5m 51s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d9b1ec38613148d0afddd5588fe46206 : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":3},{"id":"4629c018ae6c161d45cba6b394977f9bc6c41e90","author":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"date":"2026-06-03 19:30:02.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"e971c44eed985df66e64be183d9d879ffe5105a5","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-03 19:45:36.000000000","message":"Patch Set 3: Code-Review+2\n\nWe can approve this once Scott confirms.","accounts_in_message":[],"_revision_number":3},{"id":"29568cc8dc89897779e0f35862f1f6fa4adf1f4f","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2026-06-03 20:19:58.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2caf4198639dbffa54eaf8202be760655bea66bb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-03 20:39:43.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9406074be503d1bbdfb7f289d96e3ad93bfe0b78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 20:39:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b47ee019d156dcf6f85429cd114d80f8394d2ec1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 20:45:13.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fc0058a58e74f6fb3e5476a03a131a6\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8744bf13478b4ed39ced7bcfd7942245 : SUCCESS in 4m 11s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c7553faffd1e4f148bd26e24e3831cb4 : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":3},{"id":"2d445c63e0534b54e7589e36904d056be51d2f6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 20:45:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bd3d18828250144ae8fba0c22dccb9e094b5bb74","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-03 20:59:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcef8d5659b14d6791d5e8b4f4c8e593\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/2d9d32c7a45d42ce85c77b5d4abfd0fb : SUCCESS in 13m 36s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/f7e8f4edd6554162a9e09723affab685 : SUCCESS in 4m 48s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/dafbc9609d4146438d94e5c5479603c2 : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":3},{"id":"8547005a5ea47ee0b17a48986f86820639d015ec","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-03 21:00:08.000000000","message":"Patch Set 3:\n\nI\u0027ve added Scott as the initial member of the starlingx-oauth2-proxy-app-core group now.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b6816a8599c1185a688b58ac882083ddc5fa1280","revisions":{"269285843a02ac89b4d2e6eee730496a6aa58ff9":{"kind":"REWORK","_number":1,"created":"2026-06-03 15:38:11.000000000","uploader":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"ref":"refs/changes/38/991438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/991438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/991438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/991438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/991438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/991438/1"}}},"commit":{"parents":[{"commit":"5ce78e502014c2afa3835eefc32bf1812cf3ba21","subject":"Merge \"Change ACLs for Venus to retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ce78e502014c2afa3835eefc32bf1812cf3ba21"}]}],"author":{"name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","date":"2026-06-03 15:37:40.000000000","tz":-180},"committer":{"name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","date":"2026-06-03 15:38:09.000000000","tz":-180},"subject":"Add repo ouath2-proxy-app for starlingx","message":"Add repo ouath2-proxy-app for starlingx\n\nChange-Id: I30a3df6bf47be9f952427c9730cbaa6971b00122\nSigned-off-by: Italo Lemos \u003cItalo.doRegoLemos@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/269285843a02ac89b4d2e6eee730496a6aa58ff9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/269285843a02ac89b4d2e6eee730496a6aa58ff9"}]},"branch":"refs/heads/master"},"3d28c47e2b57487773d0ce366d830cf3fb169d06":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-03 15:40:00.000000000","uploader":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"ref":"refs/changes/38/991438/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/991438/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/991438/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/991438/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/991438/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/991438/2"}}},"commit":{"parents":[{"commit":"5ce78e502014c2afa3835eefc32bf1812cf3ba21","subject":"Merge \"Change ACLs for Venus to retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ce78e502014c2afa3835eefc32bf1812cf3ba21"}]}],"author":{"name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","date":"2026-06-03 15:37:40.000000000","tz":-180},"committer":{"name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","date":"2026-06-03 15:40:00.000000000","tz":0},"subject":"Add repo ouath2-proxy-app for starlingx","message":"Add repo ouath2-proxy-app for starlingx\n\nThis commit adds oauth2-proxy-app repo to StarlingX.\n\nStory: 2011810\nTask: 54757\n\nChange-Id: I30a3df6bf47be9f952427c9730cbaa6971b00122\nSigned-off-by: Italo Lemos \u003cItalo.doRegoLemos@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d28c47e2b57487773d0ce366d830cf3fb169d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/3d28c47e2b57487773d0ce366d830cf3fb169d06"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b6816a8599c1185a688b58ac882083ddc5fa1280":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2026-06-03 15:44:33.000000000","uploader":{"_account_id":36727,"name":"Italo Lemos","display_name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","username":"idoregol"},"ref":"refs/changes/38/991438/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/38/991438/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/991438/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/991438/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/38/991438/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/38/991438/3"}}},"commit":{"parents":[{"commit":"5ce78e502014c2afa3835eefc32bf1812cf3ba21","subject":"Merge \"Change ACLs for Venus to retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ce78e502014c2afa3835eefc32bf1812cf3ba21"}]}],"author":{"name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","date":"2026-06-03 15:37:40.000000000","tz":-180},"committer":{"name":"Italo Lemos","email":"Italo.doRegoLemos@windriver.com","date":"2026-06-03 15:44:33.000000000","tz":0},"subject":"Add repo oauth2-proxy-app for starlingx","message":"Add repo oauth2-proxy-app for starlingx\n\nThis commit adds oauth2-proxy-app repo to StarlingX.\n\nStory: 2011810\nTask: 54757\n\nChange-Id: I30a3df6bf47be9f952427c9730cbaa6971b00122\nSigned-off-by: Italo Lemos \u003cItalo.doRegoLemos@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b6816a8599c1185a688b58ac882083ddc5fa1280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b6816a8599c1185a688b58ac882083ddc5fa1280"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
