)]}'
{"id":"openstack%2Fproject-config~991671","triplet_id":"openstack%2Fproject-config~master~I87c2069da6c2eedbe6f714368ff9cef0c26a8c05","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"28676":{"account":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"last_update":"2026-06-05 18:53:01.000000000","reason":"Change was submitted"},"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2026-06-05 14:27:35.000000000","reason":"\u003cGERRIT_ACCOUNT_28410\u003e replied on the change","reason_account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}},"37323":{"account":{"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},"last_update":"2026-06-05 18:53:01.000000000","reason":"Change was submitted"},"31292":{"account":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"last_update":"2026-06-05 14:02:18.000000000","reason":"\u003cGERRIT_ACCOUNT_31292\u003e replied on the change","reason_account":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"}},"28558":{"account":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"last_update":"2026-06-05 18:53:01.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-06-04 21:00:29.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I87c2069da6c2eedbe6f714368ff9cef0c26a8c05","subject":"Add repo app-reloader for StarlingX","status":"MERGED","created":"2026-06-04 14:44:07.000000000","updated":"2026-06-08 12:11:40.000000000","submitted":"2026-06-05 18:53:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"991671","meta_rev_id":"89e4f29dba2feac256f85dacdf8474e48b9d0377","_number":991671,"virtual_id_number":991671,"owner":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-05 18:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":1,"date":"2026-06-05 14:27:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":1,"date":"2026-06-08 12:11:40.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-04 21:00:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"date":"2026-06-05 18:42:03.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-05 18:43:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},{"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-04 14:49:56.000000000","updated_by":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"reviewer":{"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},"state":"REVIEWER"},{"updated":"2026-06-04 14:49:56.000000000","updated_by":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"reviewer":{"_account_id":31292,"name":"ayyappa","email":"ayyappa.mantri@windriver.com","username":"amantri"},"state":"REVIEWER"},{"updated":"2026-06-04 14:49:56.000000000","updated_by":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"reviewer":{"_account_id":28558,"name":"Ghada Khalil","email":"ghada.khalil@windriver.com","username":"gkhalil"},"state":"REVIEWER"},{"updated":"2026-06-04 14:49:56.000000000","updated_by":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"},{"updated":"2026-06-04 14:49:56.000000000","updated_by":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-06-04 14:55:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-05 16:52:40.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"6bef5f30db7b8c3a6806d5aafef4c0e32afa6ca8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"date":"2026-06-04 14:44:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"09bebffdbf212325ca7078b5f2229cf1652f8127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"date":"2026-06-04 14:47:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9421de11680c5a2293beec93b1140650df68df60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 14:55:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/72defe68ba5247d6b2b1093029c15fd9\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8a6ac4a10e65486f880583bb738ff40e : SUCCESS in 5m 40s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/e61d52ee3b194406b42bb35d2065bf88 : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":2},{"id":"f1a994381f30249604794caf2edc6055f49bce61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"date":"2026-06-04 19:16:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"c9f5fc079e52d1c31039a158e332ec6772d6f238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 19:22:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f314ee0609646d686cce47c8188569d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d0e0bdf779db4431adc028b44360732b : SUCCESS in 3m 32s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/4142bd858d00414c9f21dc9d8ab5e1d7 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":3},{"id":"8e3b48f3d73a9ae10690a15ed6538b5cbc7e1a4e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-04 21:00:29.000000000","message":"Patch Set 3: Code-Review+2\n\nWe can merge this once Scott confirms.","accounts_in_message":[],"_revision_number":3},{"id":"a98830d58aedd242a857e0630d817978cdab14d4","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2026-06-05 14:27:35.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d508bbf9955e12161a9db0acf49c13f48c7e0db8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-05 15:14:03.000000000","message":"Patch Set 3:\n\nHolding since we\u0027re on the runway to a Gerrit upgrade maintenance starting in 45 minutes, but I\u0027ll approve this change once everything\u0027s back up and confirmed working.","accounts_in_message":[],"_revision_number":3},{"id":"a092a8031b7109642d2795cbfb96d16bd91040c8","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-06-05 16:52:40.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a22ef6299da749655cdabe177cd1947bb48c9795","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-06-05 18:42:03.000000000","message":"Patch Set 3: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9529ff9cf19757387a21e90cce0d132ae0dbf271","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-05 18:43:30.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"88355ac2fe614d3c996fd92fe4d9ebdecdc95d75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 18:43:42.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"af69740a9aa894083c56e97f750a0a7a499fa62c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 18:53:00.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fff19aeee704aa3964ac740318fb139\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/bcd12d207262482d8042b8a29db45129 : SUCCESS in 4m 26s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/039b1378d8124be18294b46dc470835e : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":3},{"id":"cb2742a6b01ab32c0e46d28c04131561bdfc11aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 18:53:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a6d94babaaff038c46d26bf43c310273fa7d242d","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-05 19:05:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68b613bf319642b6ba48beb11c65a007\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/59d14fb94d124e9dba15597aed1ed3ef : SUCCESS in 12m 01s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/2ff771beb846484a8a8ae7a15780f985 : SUCCESS in 4m 33s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/ee3088de23fe495bbe1cd2d5325f5e97 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":3},{"id":"bc15b2bf0dab2e652209a886e1b1cce2680b2b93","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-05 19:09:56.000000000","message":"Patch Set 3:\n\nI\u0027ve added Scott as the initial member of the starlingx-app-reloader-core group now.","accounts_in_message":[],"_revision_number":3},{"id":"89e4f29dba2feac256f85dacdf8474e48b9d0377","author":{"_account_id":37323,"name":"Eduardo Almeida dos Santos","display_name":"Eduardo Almeida","email":"Eduardo.AlmeidadosSantos@windriver.com","username":"ealmeida"},"date":"2026-06-08 12:11:40.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a002be678fe14322d7a44933bb0c0cbbf0ddca71","revisions":{"a8b73a1f88d6d573e853cefdc99943c5202fd6e6":{"kind":"REWORK","_number":1,"created":"2026-06-04 14:44:07.000000000","uploader":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"ref":"refs/changes/71/991671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/71/991671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/991671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/991671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/991671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/71/991671/1"}}},"commit":{"parents":[{"commit":"5ce78e502014c2afa3835eefc32bf1812cf3ba21","subject":"Merge \"Change ACLs for Venus to retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ce78e502014c2afa3835eefc32bf1812cf3ba21"}]}],"author":{"name":"Andy Ning","email":"andy.ning@windriver.com","date":"2026-06-04 14:40:51.000000000","tz":-240},"committer":{"name":"Andy Ning","email":"andy.ning@windriver.com","date":"2026-06-04 14:40:51.000000000","tz":-240},"subject":"Add repo app-reloader for StarlingX","message":"Add repo app-reloader for StarlingX\n\nThis commit adds a new repo named app-reloader, an a Kubernetes\napp that automatically triggers rollouts of workloads (like Deployments,\nStatefulSets, and more) whenever referenced Secrets, ConfigMaps or\noptionally CSI-mounted secrets are updated [1]. It\u0027ll be used to monitor\nand auto renew TLS certificate for EJBCA service endpoint.\n\n[1]: https://github.com/stakater/Reloader\n\nChange-Id: I87c2069da6c2eedbe6f714368ff9cef0c26a8c05\nSigned-off-by: Andy Ning \u003candy.ning@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a8b73a1f88d6d573e853cefdc99943c5202fd6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a8b73a1f88d6d573e853cefdc99943c5202fd6e6"}]},"branch":"refs/heads/master"},"ed7db09cef953dbe1983caefd404359e0f62949e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-04 14:47:54.000000000","uploader":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"ref":"refs/changes/71/991671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/71/991671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/991671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/991671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/991671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/71/991671/2"}}},"commit":{"parents":[{"commit":"5ce78e502014c2afa3835eefc32bf1812cf3ba21","subject":"Merge \"Change ACLs for Venus to retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ce78e502014c2afa3835eefc32bf1812cf3ba21"}]}],"author":{"name":"Andy Ning","email":"andy.ning@windriver.com","date":"2026-06-04 14:40:51.000000000","tz":-240},"committer":{"name":"Andy Ning","email":"andy.ning@windriver.com","date":"2026-06-04 14:46:45.000000000","tz":-240},"subject":"Add repo app-reloader for StarlingX","message":"Add repo app-reloader for StarlingX\n\nThis commit adds a new repo named app-reloader, an a Kubernetes\napp that automatically triggers rollouts of workloads (like Deployments,\nStatefulSets, and more) whenever referenced Secrets, ConfigMaps or\noptionally CSI-mounted secrets are updated [1]. It\u0027ll be used to monitor\nand auto renew TLS certificate for EJBCA service endpoint.\n\n[1]: https://github.com/stakater/Reloader\n\nStory: 2011712\nTask: 54763\n\nChange-Id: I87c2069da6c2eedbe6f714368ff9cef0c26a8c05\nSigned-off-by: Andy Ning \u003candy.ning@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ed7db09cef953dbe1983caefd404359e0f62949e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ed7db09cef953dbe1983caefd404359e0f62949e"}]},"branch":"refs/heads/master"},"a002be678fe14322d7a44933bb0c0cbbf0ddca71":{"kind":"REWORK","_number":3,"created":"2026-06-04 19:16:32.000000000","uploader":{"_account_id":28676,"name":"Andy Ning","email":"andy.ning@windriver.com","username":"andy.wrs"},"ref":"refs/changes/71/991671/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/71/991671/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/991671/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/991671/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/71/991671/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/71/991671/3"}}},"commit":{"parents":[{"commit":"5ce78e502014c2afa3835eefc32bf1812cf3ba21","subject":"Merge \"Change ACLs for Venus to retired\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5ce78e502014c2afa3835eefc32bf1812cf3ba21"}]}],"author":{"name":"Andy Ning","email":"andy.ning@windriver.com","date":"2026-06-04 14:40:51.000000000","tz":-240},"committer":{"name":"Andy Ning","email":"andy.ning@windriver.com","date":"2026-06-04 19:16:05.000000000","tz":-240},"subject":"Add repo app-reloader for StarlingX","message":"Add repo app-reloader for StarlingX\n\nThis commit adds a new repo named app-reloader, an a Kubernetes\napp that automatically triggers rollouts of workloads (like Deployments,\nStatefulSets, and more) whenever referenced Secrets, ConfigMaps or\noptionally CSI-mounted secrets are updated [1]. It\u0027ll be used to monitor\nand auto renew TLS certificate for EJBCA service endpoint.\n\n[1]: https://github.com/stakater/Reloader\n\nStory: 2011712\nTask: 54763\n\nChange-Id: I87c2069da6c2eedbe6f714368ff9cef0c26a8c05\nSigned-off-by: Andy Ning \u003candy.ning@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a002be678fe14322d7a44933bb0c0cbbf0ddca71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a002be678fe14322d7a44933bb0c0cbbf0ddca71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
