)]}'
{"id":"openstack%2Fproject-config~993184","triplet_id":"openstack%2Fproject-config~master~I5c60e006963472590c101d9e3485f6db75468940","project":"openstack/project-config","branch":"master","topic":"retire-cinderlib","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2026-09-01 14:21:48.000000000","reason":"Change was submitted"},"9236":{"account":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"last_update":"2026-09-01 14:21:48.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-06-25 08:06:23.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2026-09-01 14:21:48.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2026-09-01 14:21:48.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-09-01 14:11:33.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I5c60e006963472590c101d9e3485f6db75468940","subject":"Retire cinderlib","status":"MERGED","created":"2026-06-12 20:57:19.000000000","updated":"2026-09-01 14:30:32.000000000","submitted":"2026-09-01 14:21:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"993184-retire-cinderlib","meta_rev_id":"ec84808ad1bf78b1de8a58728d7263361ebacb68","_number":993184,"virtual_id_number":993184,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-09-01 14:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2026-09-01 06:44:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-09-01 14:11:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-09-01 14:11:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-12 21:09:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-15 14:44:50.000000000","updated_by":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"reviewer":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"state":"REVIEWER"},{"updated":"2026-06-17 14:43:25.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-06-19 05:36:17.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-06-19 05:36:17.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2026-06-23 11:21:16.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"}],"messages":[{"id":"3c9d63c9386876bca65b10d36a59b75b752d9d9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-12 20:57:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cff333b44bddc0a03befb137b1f4d24d850227c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-12 21:09:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a772bad16d054a1f9c17334163ae871d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1a1f425fa5ff4aabb5978cf2984d178b : SUCCESS in 3m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/a4ff4ef66b164a2790c9ee9192424c62 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":1},{"id":"98c1a8f9735b954eef6038ba953a4b423c3b6d61","author":{"_account_id":9236,"name":"Jon Bernard","email":"jobernar@redhat.com","username":"jbernard"},"date":"2026-06-15 14:44:50.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a4a4f67d60ac44a53e7c77019d651996a2b3e5f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-17 14:43:25.000000000","message":"Patch Set 1: Code-Review+1\n\nI think you\u0027ll want to ask the unmaintainers to delete the remaining open unmaintained branches before merging this ACL change, so I\u0027m holding off my +2 to get clarification on what the plan is there.\n\nAlso, not a blocker, but don\u0027t forget to retire the deliverable in openstack/governance as well (it\u0027s still listed as deprecated there).","accounts_in_message":[],"_revision_number":1},{"id":"ad8b37dac4280bae084a91ac30531962d34de230","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-18 19:47:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66b73ffb2d0ae4dd885b0434e9d70e5955bb94ea","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-18 22:01:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5565a738b8aa38c2abe64007f2a5ced12ac8088a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-19 05:36:17.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a13c987cc1d8ad9ca38b3786aa3321df38d54c3d","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2026-06-23 11:21:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76ffac1c756392b931ae7332f35fd09a5f9001e9","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-23 23:07:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c19ba53a6f34e9ec46aac7bdb2f73fc6fde0eab8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-24 12:43:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5eafffd1980593461167127b4c8653c0c277c5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-24 12:44:35.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"fe8418210f9f4b33dc13d2a8921a80d58aefb100","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 12:46:30.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/charm-openstack-hypervisor\" was not found.  All\n  projects referenced within a Zuul configuration must first be added to\n  the main configuration file by the Zuul administrator.\n\nThe problem appears in the \"openstack/charm-openstack-hypervisor\" project stanza:\n\n  project:\n      # Retired Sunbeam single charm repositories\n      name: openstack/charm-openstack-hypervisor\n      templates:\n        - official-openstack-repo-jobs\n        - noop-jobs\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 1492","accounts_in_message":[],"_revision_number":2},{"id":"5fa01ef8d0bb9060b5673408be6a8a3b174498a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-24 12:50:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"411c6c9073680e8b993ce5861ea0688bdbc89e60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 12:51:50.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/project-config on branch master.  The\nproblem was:\n\n  The project \"openstack/charm-openstack-hypervisor\" was not found.  All\n  projects referenced within a Zuul configuration must first be added to\n  the main configuration file by the Zuul administrator.\n\nThe problem appears in the \"openstack/charm-openstack-hypervisor\" project stanza:\n\n  project:\n      # Retired Sunbeam single charm repositories\n      name: openstack/charm-openstack-hypervisor\n      templates:\n        - official-openstack-repo-jobs\n        - noop-jobs\n\n  in \"openstack/project-config/zuul.d/projects.yaml@master\", line 1492","accounts_in_message":[],"_revision_number":3},{"id":"e9943d059f904cced9cbcaf7b7acf69fd65bb13a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-24 13:03:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"36b95aebe1b07e05572dcdbb16362156b95f76c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2026-06-24 19:23:53.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"e0f04c96c2fe744d221f19efabd6ccf0a0990c8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 19:31:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c639e8ffd5e46a090184372b9ca2e33\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a8bc1d89b6134948a1c57be124d7ff44 : SUCCESS in 5m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/2e41df3b291b48eeb948011ca5946cf5 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":4},{"id":"6ad6760cd779ce2717da1e5dc66661ac9ac85c8c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-25 08:06:23.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a7447207bdadc22597f90272d0cef909803db052","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-09-01 06:44:37.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f85e056ebbb8e962c0d86a8f70248027c385d2e0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-09-01 14:11:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nConfirmed, Gitea reports no branches other than master which contains just the readme, Gerrit indicates there are no remaining changes open.","accounts_in_message":[],"_revision_number":4},{"id":"50f7dc22083176801cbfbbd855425fab48277a3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:12:11.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8c8c6e2cffd2b27e9d6672f73b707033284a9f5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:21:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6edb4df9d4849ec84754cb1e99fc870\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1ac2440859b1489a88ed7a10fb9c250c : SUCCESS in 4m 56s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/0890a7b9117e4177b5f1738a3388ebf2 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":4},{"id":"069ed8093c94334c811a84c409aaf54e02deb27f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:21:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ec84808ad1bf78b1de8a58728d7263361ebacb68","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-01 14:30:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4eded353d9a64b269fa4ceaa48b29da2\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/de15ed397a7745c0ab02bf5b509ce941 : SUCCESS in 7m 44s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/0bc37a6c370f48468de12e0a31aa39d0 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d996eed9fbfe86c9860d3ffd1aa3df3374f89996","revisions":{"809539a7a56db1bdf0dbf83023031e91bd88b158":{"kind":"REWORK","_number":1,"created":"2026-06-12 20:57:19.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/84/993184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/84/993184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/84/993184/1"}}},"commit":{"parents":[{"commit":"c20cbf1c6a1120624c962a467634c4b5f0565fa2","subject":"Merge \"Add 3 more Metal3 projects to untrusted-projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c20cbf1c6a1120624c962a467634c4b5f0565fa2"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-06-12 20:55:55.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-06-12 20:56:58.000000000","tz":-240},"subject":"Retire cinderlib","message":"Retire cinderlib\n\nThe cinderlib library was DEPRECATED in December 2023 [0] with a\ncommitment from the cinder project to maintain the stable branches\neven though development had stopped on the project [1].  The most\nrecent stable branch was stable/2023.2 (Bobcat), which went EOL in\nApril 2025 [2].  Therefore, it is time to declare cinderlib RETIRED.\n\n[0] change Ib186ac5830e5920e264d79be946995e63e960426\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/L6HJW55SEUL4NYQVOESJ22KFDW5SGZAE/\n[2] change Ib3ebf5f918674a48baf6650be1e06181e420b354\n\nChange-Id: I5c60e006963472590c101d9e3485f6db75468940\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/809539a7a56db1bdf0dbf83023031e91bd88b158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/809539a7a56db1bdf0dbf83023031e91bd88b158"}]},"branch":"refs/heads/master"},"33b47ab6e4502824b44b9add384563f7a69d989e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2026-06-24 12:44:35.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/84/993184/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/84/993184/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/84/993184/2"}}},"commit":{"parents":[{"commit":"c20cbf1c6a1120624c962a467634c4b5f0565fa2","subject":"Merge \"Add 3 more Metal3 projects to untrusted-projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c20cbf1c6a1120624c962a467634c4b5f0565fa2"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-06-12 20:55:55.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-06-24 12:44:35.000000000","tz":0},"subject":"Retire cinderlib","message":"Retire cinderlib\n\nThe cinderlib library was DEPRECATED in December 2023 [0] with a\ncommitment from the cinder project to maintain the stable branches\neven though development had stopped on the project [1].  The most\nrecent stable branch was stable/2023.2 (Bobcat), which went EOL in\nApril 2025 [2].  Therefore, it is time to declare cinderlib RETIRED.\n\n[0] change Ib186ac5830e5920e264d79be946995e63e960426\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/L6HJW55SEUL4NYQVOESJ22KFDW5SGZAE/\n[2] change Ib3ebf5f918674a48baf6650be1e06181e420b354\n\nDepends-on: https://review.opendev.org/c/openstack/releases/+/994570\n\nChange-Id: I5c60e006963472590c101d9e3485f6db75468940\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/33b47ab6e4502824b44b9add384563f7a69d989e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/33b47ab6e4502824b44b9add384563f7a69d989e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ea4c31eee6e61245e31f621afd2523e836193e25":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-06-24 12:50:52.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/84/993184/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/84/993184/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/84/993184/3"}}},"commit":{"parents":[{"commit":"6d4b1f195213c895d60565a92920a5761d3adc2a","subject":"Merge \"sunbeam: retire all single charm repositories\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6d4b1f195213c895d60565a92920a5761d3adc2a"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-06-12 20:55:55.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-06-24 12:50:52.000000000","tz":0},"subject":"Retire cinderlib","message":"Retire cinderlib\n\nThe cinderlib library was DEPRECATED in December 2023 [0] with a\ncommitment from the cinder project to maintain the stable branches\neven though development had stopped on the project [1].  The most\nrecent stable branch was stable/2023.2 (Bobcat), which went EOL in\nApril 2025 [2].  Therefore, it is time to declare cinderlib RETIRED.\n\n[0] change Ib186ac5830e5920e264d79be946995e63e960426\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/L6HJW55SEUL4NYQVOESJ22KFDW5SGZAE/\n[2] change Ib3ebf5f918674a48baf6650be1e06181e420b354\n\nDepends-on: https://review.opendev.org/c/openstack/releases/+/994570\n\nChange-Id: I5c60e006963472590c101d9e3485f6db75468940\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ea4c31eee6e61245e31f621afd2523e836193e25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ea4c31eee6e61245e31f621afd2523e836193e25"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"c20cbf1c6a1120624c962a467634c4b5f0565fa2","ours":"33b47ab6e4502824b44b9add384563f7a69d989e","theirs":"6d4b1f195213c895d60565a92920a5761d3adc2a","merge_strategy":"recursive","contains_conflicts":false}},"d996eed9fbfe86c9860d3ffd1aa3df3374f89996":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-06-24 19:23:53.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/84/993184/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/84/993184/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/84/993184/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/84/993184/4"}}},"commit":{"parents":[{"commit":"eaf841ea7f0155d0c782c0baa37cf56f10dcb9e8","subject":"Drop config for retired Sunbeam single charm repositories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/eaf841ea7f0155d0c782c0baa37cf56f10dcb9e8"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-06-12 20:55:55.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2026-06-24 19:23:53.000000000","tz":0},"subject":"Retire cinderlib","message":"Retire cinderlib\n\nThe cinderlib library was DEPRECATED in December 2023 [0] with a\ncommitment from the cinder project to maintain the stable branches\neven though development had stopped on the project [1].  The most\nrecent stable branch was stable/2023.2 (Bobcat), which went EOL in\nApril 2025 [2].  Therefore, it is time to declare cinderlib RETIRED.\n\n[0] change Ib186ac5830e5920e264d79be946995e63e960426\n[1] https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/L6HJW55SEUL4NYQVOESJ22KFDW5SGZAE/\n[2] change Ib3ebf5f918674a48baf6650be1e06181e420b354\n\nDepends-on: https://review.opendev.org/c/openstack/releases/+/994570\n\nChange-Id: I5c60e006963472590c101d9e3485f6db75468940\nSigned-off-by: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d996eed9fbfe86c9860d3ffd1aa3df3374f89996"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/d996eed9fbfe86c9860d3ffd1aa3df3374f89996"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"6d4b1f195213c895d60565a92920a5761d3adc2a","ours":"ea4c31eee6e61245e31f621afd2523e836193e25","theirs":"eaf841ea7f0155d0c782c0baa37cf56f10dcb9e8","merge_strategy":"recursive","contains_conflicts":false}}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
