)]}'
{"id":"openstack%2Fproject-config~993214","triplet_id":"openstack%2Fproject-config~master~I169b37721a145673c50cb6e263ff7453aebfd5f5","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"2":{"account":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"last_update":"2026-06-17 17:37:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I169b37721a145673c50cb6e263ff7453aebfd5f5","subject":"Add more repos for drizzle website and sysadmin automation","status":"MERGED","created":"2026-06-13 13:40:16.000000000","updated":"2026-06-17 17:49:35.000000000","submitted":"2026-06-17 17:37:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"993214","meta_rev_id":"812e6907d7e77f982b6d770fb8e3722e85f5ccec","_number":993214,"virtual_id_number":993214,"owner":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-17 17:37:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":2,"date":"2026-06-17 15:30:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-17 17:28:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-17 17:28:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-13 13:44:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-17 15:30:51.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2026-06-17 17:28:03.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"9fe694c49a37f95abc08ab78e00801db7fad6edb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-13 13:40:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afdb035dc68eacf5639099fd734df18281975d46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-13 13:44:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11aadc334e3347949e19bcdc65365f4e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f6a9fd28fb9941069e4ccdfac8e53808 : SUCCESS in 3m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c8a1c36364cf4959808df4ca67f4ab2e : FAILURE in 3m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8142207a27ed1a6346a894da68ff1398e636ff02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-15 15:35:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"7cbe342bdabd1c097a471b591d6226d5b5fc2409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-15 15:45:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec6dad956d6f42b99b656a9f92e4eb6e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/f7e9318c5b624e70aa2e8c34595e4251 : SUCCESS in 4m 43s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/fa7539a8db4e483bbcaeed168059057c : SUCCESS in 2m 56s","accounts_in_message":[],"_revision_number":2},{"id":"38f40132cc81a7f2d5dd56d983fef70521b7d41a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2026-06-17 15:17:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"d11f72b5df3c19adebef7d90f65c70b95c8a2cd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 15:26:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/883e86e498d244d4b7f69c6c7b59753e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d122853d535d4871bce288831a1575c4 : SUCCESS in 4m 42s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/d4248f57f5fa406b94c3c7c060b0d8eb : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":3},{"id":"37e4fc38846476b42a28985f47596c21ccfcca79","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2026-06-17 15:30:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"76cd8650e35d77ddfca9a913be05887531c99c41","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-17 17:28:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"569178d269049686418bfe9844b1dfd20b84a46f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 17:28:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"814c8ab9a1c8f00b6afcf30cd8aaa3d6b16fa0c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 17:37:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c9ef6a808014747b0d6211d923c73ef\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/121bb3c556d847e789ebfea650e17de5 : SUCCESS in 4m 25s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/76010aa710ac4c7083381e0c3a3b423c : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"686608294360d88ba24ca272b5553c568bd886db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 17:37:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"812e6907d7e77f982b6d770fb8e3722e85f5ccec","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-17 17:49:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad5096d44e8c4508a8963d9217d504b6\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/03fc4e0d3c9d4a0c904878916b972a14 : SUCCESS in 11m 48s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/32fa66ef721a463791f4fac7a44ea952 : SUCCESS in 4m 37s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/48bde2c94ad249cda84f1eeebd14ceea : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"71cc7ff8e18600d47822b00fe6c558f436e8be21","revisions":{"be2733e844775c5bda611c76b002f51138732f70":{"kind":"REWORK","_number":1,"created":"2026-06-13 13:40:16.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/14/993214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/993214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/993214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/993214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/993214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/993214/1"}}},"commit":{"parents":[{"commit":"c20cbf1c6a1120624c962a467634c4b5f0565fa2","subject":"Merge \"Add 3 more Metal3 projects to untrusted-projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c20cbf1c6a1120624c962a467634c4b5f0565fa2"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-13 13:13:15.000000000","tz":0},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-13 13:13:15.000000000","tz":0},"subject":"Add repos for drizzle website and sysadmin automation","message":"Add repos for drizzle website and sysadmin automation\n\nWe got re-approved by the SPI so we have our domain name back. We\u0027re\ngoing to start launching a few things, including a website and a\nperf regression tracking db (running drizzle, natch) Possible our\nown dedicated hermes agents? Anyway, we need a couple of more repos.\n\nChange-Id: I169b37721a145673c50cb6e263ff7453aebfd5f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be2733e844775c5bda611c76b002f51138732f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/be2733e844775c5bda611c76b002f51138732f70"}]},"branch":"refs/heads/master"},"773ccdba9b2adf990784723b2a2ab1e359aa36e0":{"kind":"REWORK","_number":2,"created":"2026-06-15 15:35:23.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/14/993214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/993214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/993214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/993214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/993214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/993214/2"}}},"commit":{"parents":[{"commit":"c20cbf1c6a1120624c962a467634c4b5f0565fa2","subject":"Merge \"Add 3 more Metal3 projects to untrusted-projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c20cbf1c6a1120624c962a467634c4b5f0565fa2"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-13 13:13:15.000000000","tz":0},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-15 15:35:14.000000000","tz":0},"subject":"Add repos for drizzle website and sysadmin automation","message":"Add repos for drizzle website and sysadmin automation\n\nWe got re-approved by the SPI so we have our domain name back. We\u0027re\ngoing to start launching a few things, including a website and a\nperf regression tracking db (running drizzle, natch) Possible our\nown dedicated hermes agents? Anyway, we need a couple of more repos.\n\nChange-Id: I169b37721a145673c50cb6e263ff7453aebfd5f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/773ccdba9b2adf990784723b2a2ab1e359aa36e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/773ccdba9b2adf990784723b2a2ab1e359aa36e0"}]},"branch":"refs/heads/master"},"71cc7ff8e18600d47822b00fe6c558f436e8be21":{"kind":"REWORK","_number":3,"created":"2026-06-17 15:17:21.000000000","uploader":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"ref":"refs/changes/14/993214/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/14/993214/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/993214/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/993214/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/14/993214/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/14/993214/3"}}},"commit":{"parents":[{"commit":"c20cbf1c6a1120624c962a467634c4b5f0565fa2","subject":"Merge \"Add 3 more Metal3 projects to untrusted-projects\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c20cbf1c6a1120624c962a467634c4b5f0565fa2"}]}],"author":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-13 13:13:15.000000000","tz":0},"committer":{"name":"Monty Taylor","email":"mordred@inaugust.com","date":"2026-06-17 15:17:05.000000000","tz":0},"subject":"Add more repos for drizzle website and sysadmin automation","message":"Add more repos for drizzle website and sysadmin automation\n\nWe got re-approved by the SPI so we have our domain name back. We\u0027re\ngoing to start launching a few things, including a website and a\nperf regression tracking db (running drizzle, natch) Possibly our\nown dedicated hermes agents? Anyway, we need a couple of more repos.\n\nWe\u0027re also exploring a consolidated agent interface via MCP rather\nthan explicit plugins for openclaw and hermes, neither of which can\nbe used by claude on the web. That\u0027s not really a drizzle project.\n\nChange-Id: I169b37721a145673c50cb6e263ff7453aebfd5f5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/71cc7ff8e18600d47822b00fe6c558f436e8be21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/71cc7ff8e18600d47822b00fe6c558f436e8be21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
