)]}'
{"id":"openstack%2Fproject-config~994192","triplet_id":"openstack%2Fproject-config~master~I2aa59794e6b4b345bf1967bcc27723f765a69ee1","project":"openstack/project-config","branch":"master","topic":"projects-zuul-cleanup","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-06-22 08:42:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2aa59794e6b4b345bf1967bcc27723f765a69ee1","subject":"Drop x/networking-bigswitch from zuul","status":"MERGED","created":"2026-06-20 19:52:28.000000000","updated":"2026-06-22 08:49:06.000000000","submitted":"2026-06-22 08:42:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994192-projects-zuul-cleanup","meta_rev_id":"e93af0026dee9d2c4a1f0c28ad96b5ba33f61097","_number":994192,"virtual_id_number":994192,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-22 08:42:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-06-22 08:32:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-06-22 08:32:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-20 20:06:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-22 08:32:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"8623ed832fcd9ffc50f8c40d8adfd4ef91dbf988","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-20 19:52:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fe3687795e730e0afd4dc9bb43c02e35d44c03a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-20 20:06:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9aeb7841cee4cb4b166a4f3849a0592\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/de667c6ddc9a4b0c8fcaf2a440b6b918 : SUCCESS in 4m 22s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/5ca4f4c8dd014bfdb9b38f479771ff88 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4f4433aa1779bb4a44de2af1b43677cd3850d86f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-22 08:32:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43c5a785230af66820418f35717c8608bc9f678a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 08:32:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d324c9d8e459f3c03f9579135e0bbf67555ee51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 08:42:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dba1854afbcc435f85728f2f5244e847\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/1119e86b08ff471296ce6ce974539da9 : SUCCESS in 3m 04s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/c722d9f73a5a4afdac1d1b35f6644bdc : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b363fb42cc879715746ceede65028aef9902d0ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 08:42:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e93af0026dee9d2c4a1f0c28ad96b5ba33f61097","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-22 08:49:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27eff07e308b4a4cbbf0a6200b6ee6c2\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/4c1ead4646fa48ecac65881d66ba2d26 : SUCCESS in 6m 30s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/51ef4f40042443588bdf602c4c366937 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e27083f6e84feeb40d8775cdd86a859070323c5","revisions":{"5e27083f6e84feeb40d8775cdd86a859070323c5":{"kind":"REWORK","_number":1,"created":"2026-06-20 19:52:28.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/92/994192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/994192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/994192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/994192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/994192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/994192/1"}}},"commit":{"parents":[{"commit":"8034e65d56b212f2dfc7a17af07a705f4e52aaff","subject":"Drop x/networking-bigswitch from zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/8034e65d56b212f2dfc7a17af07a705f4e52aaff"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-20 19:51:18.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-20 19:51:54.000000000","tz":120},"subject":"Drop x/networking-bigswitch from zuul","message":"Drop x/networking-bigswitch from zuul\n\nMissed to drop the second reference in [0].\n\n[0] I9caddb292b0415540c7efaef4a675c46a7af6924\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I2aa59794e6b4b345bf1967bcc27723f765a69ee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e27083f6e84feeb40d8775cdd86a859070323c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e27083f6e84feeb40d8775cdd86a859070323c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
