)]}'
{"id":"openstack%2Fproject-config~994392","triplet_id":"openstack%2Fproject-config~master~I68672c30614ea557d76cf0f23c6b40ae6d3ad87e","project":"openstack/project-config","branch":"master","topic":"projects-zuul-cleanup","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2026-06-23 10:51:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I68672c30614ea557d76cf0f23c6b40ae6d3ad87e","subject":"Drop x/neutron-interconnection from zuul","status":"MERGED","created":"2026-06-23 05:42:17.000000000","updated":"2026-06-23 11:00:27.000000000","submitted":"2026-06-23 10:51:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994392-projects-zuul-cleanup","meta_rev_id":"87adb86d9d461a1fc312c0dae7d6639bbf13f040","_number":994392,"virtual_id_number":994392,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-23 10:51:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-06-23 10:09:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-06-23 10:09:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-23 05:52:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-23 10:09:23.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ed1f3c1559998e27a1aff955e820a820cb79f617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-06-23 05:42:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae4c1fba55dbe2416249fb358e14cf51e83f89d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 05:52:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bb6a44de8d145758802d4192238792a\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/e9ece63595c5423dbd94edbcb1aceaa3 : SUCCESS in 2m 30s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/66fe897f7e694aadaefc2b5abd0ab226 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9cbb7b758921d70498aac97cfc113e642e50523f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-23 10:09:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ee7bd1c2c440fdab547dda1ced449a3cfac52d59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 10:09:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffad554dddb9de01f892d42b1f9f1a3766b57875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 10:51:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98d7533d9caa4e16aca551c665dde072\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/ff562fae1bbd499f98e91ed8e3d20833 : SUCCESS in 4m 46s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f3ebef36d3794b03b478c38c45aa8992 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"1e0fb85b373b23b1431c8ecffeb701d00f4b3c73","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 10:51:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"87adb86d9d461a1fc312c0dae7d6639bbf13f040","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 11:00:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c1ed88999054110beb363cc19c7d158\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/255d1441b9514a67be6c9a88731429df : SUCCESS in 8m 25s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/ca4dd9e5d5784dd49377d282e62ba72a : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a66fcff2144c424fc1e78c060a8c45d0ca7f141f","revisions":{"a66fcff2144c424fc1e78c060a8c45d0ca7f141f":{"kind":"REWORK","_number":1,"created":"2026-06-23 05:42:17.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/92/994392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/92/994392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/994392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/994392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/92/994392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/92/994392/1"}}},"commit":{"parents":[{"commit":"5e27083f6e84feeb40d8775cdd86a859070323c5","subject":"Drop x/networking-bigswitch from zuul","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/5e27083f6e84feeb40d8775cdd86a859070323c5"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-23 05:39:18.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2026-06-23 05:39:18.000000000","tz":120},"subject":"Drop x/neutron-interconnection from zuul","message":"Drop x/neutron-interconnection from zuul\n\nThe repos has had no activity since 2019[0].\n\nThe repo is blocking cleanup of old openstack templates[1].\n\nIt can be added to zuul again once someone starts doing the necessary\ncleanup work\n\n[0] https://review.opendev.org/q/project:x/neutron-interconnection\n[1] https://review.opendev.org/c/openstack/openstack-zuul-jobs/+/994188/1\n\nSigned-off-by: Dr. Jens Harbott \u003charbott@osism.tech\u003e\nChange-Id: I68672c30614ea557d76cf0f23c6b40ae6d3ad87e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a66fcff2144c424fc1e78c060a8c45d0ca7f141f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a66fcff2144c424fc1e78c060a8c45d0ca7f141f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
