)]}'
{"id":"openstack%2Fproject-config~994823","triplet_id":"openstack%2Fproject-config~master~I9e1175fdc8c77412c11a75b9587bc2853018c442","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2026-06-25 19:22:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e1175fdc8c77412c11a75b9587bc2853018c442","subject":"Remove branch deletion powers from three new repo","status":"MERGED","created":"2026-06-24 20:48:03.000000000","updated":"2026-06-25 22:14:34.000000000","submitted":"2026-06-25 19:22:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994823","meta_rev_id":"5cef894116181b56679ad5cf4680a8db8f789efe","_number":994823,"virtual_id_number":994823,"owner":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 19:22:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-25 19:12:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-25 19:12:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 20:59:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-25 19:12:57.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"80ebe8dec0fd0cdb04a797f1a5931ba604e641ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2026-06-24 20:48:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"799e392b69ee9c3e03e5e560d00dd7b99f40a1ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 20:59:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7dcce07b828c4a1da695e59f4c07c212\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/926068c07bc24796a8a271a48ae5c5ca : SUCCESS in 6m 14s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/483f1439e99d488aa2ed946f50810229 : FAILURE in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"89b6bdf686467b8df3073213429bc5f48c041514","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2026-06-25 19:02:12.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"8556f44f52aba4116a85e67b2b6af20d909c7552","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 19:10:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8baa546f85e4713a0dafaa80915aa8b\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d8e8683720f34f1bad5eafa69f605b21 : SUCCESS in 6m 19s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f1f153de97094e15ba7b9521deccc850 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"0661f426272876de52629ac32e5e6efb853d2746","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-25 19:12:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4d9685a7d7191c29347a4b38159c4f61fdcdb897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 19:13:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"30ef0f81b4e5f3d4fd1c162cfee8fe6b35b192b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 19:22:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eb4a063ecad48bdaa2209d28a22d4a5\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/7c992df0012f4576bb9cf060cfc052ec : SUCCESS in 5m 03s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/f5332a8236364fa6a7541f4b0a0f6bec : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":2},{"id":"d47e344e996a2405f21f6e23847eb16b9c4d8746","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 19:22:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5cef894116181b56679ad5cf4680a8db8f789efe","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 22:14:34.000000000","message":"Patch Set 2:\n\nBuild failed (deploy pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7be59684d45142a6a157a45f5811462e\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/4e3123645245472eb4aa01a8171c48c1 : SUCCESS in 1h 22m 20s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/8d5f59b45a8c4183bf55f0e1dde7b447 : TIMED_OUT in 1h 20m 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"45496c5839784c76f38c8ac7d4c38f17f22cc808","revisions":{"aa4d23f18df3226a83fd15b2d6a388a4a7706380":{"kind":"REWORK","_number":1,"created":"2026-06-24 20:48:03.000000000","uploader":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"ref":"refs/changes/23/994823/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/994823/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/994823/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/994823/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/994823/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/994823/1"}}},"commit":{"parents":[{"commit":"24ee7db2d514891aff7a9e99c478df3e81f5e572","subject":"add repo build-config for StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/24ee7db2d514891aff7a9e99c478df3e81f5e572"}]}],"author":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2026-06-24 20:34:48.000000000","tz":-240},"committer":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2026-06-24 20:34:54.000000000","tz":-240},"subject":"Remove branch deletion powers from three new repo","message":"Remove branch deletion powers from three new repo\n\nThree recently added repos repicated the branch deletion\npower of starlingx-release.  We are removing that permission.\nRemoving the permission from all other repos were covered by\nthis update ...\nhttps://review.opendev.org/c/openstack/project-config/+/981169\n\nSigned-off-by: Scott Little \u003cscott.little@windriver.com\u003e\nChange-Id: I9e1175fdc8c77412c11a75b9587bc2853018c442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aa4d23f18df3226a83fd15b2d6a388a4a7706380"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/aa4d23f18df3226a83fd15b2d6a388a4a7706380"}]},"branch":"refs/heads/master"},"45496c5839784c76f38c8ac7d4c38f17f22cc808":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-06-25 19:02:12.000000000","uploader":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"ref":"refs/changes/23/994823/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/23/994823/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/994823/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/994823/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/23/994823/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/23/994823/2"}}},"commit":{"parents":[{"commit":"c1f4b1fd5df3b3291da0a88ccf23ab46daee9a03","subject":"add repo build-config for StarlingX","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/c1f4b1fd5df3b3291da0a88ccf23ab46daee9a03"}]}],"author":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2026-06-24 20:34:48.000000000","tz":-240},"committer":{"name":"Scott Little","email":"scott.little@windriver.com","date":"2026-06-25 19:02:03.000000000","tz":-240},"subject":"Remove branch deletion powers from three new repo","message":"Remove branch deletion powers from three new repo\n\nThree recently added repos repicated the branch deletion\npower of starlingx-release.  We are removing that permission.\nRemoving the permission from all other repos were covered by\nthis update ...\nhttps://review.opendev.org/c/openstack/project-config/+/981169\n\nSigned-off-by: Scott Little \u003cscott.little@windriver.com\u003e\nChange-Id: I9e1175fdc8c77412c11a75b9587bc2853018c442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45496c5839784c76f38c8ac7d4c38f17f22cc808"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/45496c5839784c76f38c8ac7d4c38f17f22cc808"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
