)]}'
{"id":"openstack%2Fproject-config~994994","triplet_id":"openstack%2Fproject-config~master~Ic28d37f3056362697fcde97a4704a645a243581d","project":"openstack/project-config","branch":"master","topic":"new-project","attention_set":{},"removed_from_attention_set":{"32852":{"account":{"_account_id":32852,"name":"Enzo Candotti","email":"enzo.candotti@windriver.com","username":"ecandotti"},"last_update":"2026-06-29 20:35:16.000000000","reason":"Change was submitted"},"26026":{"account":{"_account_id":26026,"name":"Greg Waines","email":"greg.waines@windriver.com","username":"gwaines"},"last_update":"2026-06-29 20:35:16.000000000","reason":"Change was submitted"},"28410":{"account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"last_update":"2026-06-29 18:23:06.000000000","reason":"\u003cGERRIT_ACCOUNT_28410\u003e replied on the change","reason_account":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-06-26 16:17:37.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic28d37f3056362697fcde97a4704a645a243581d","subject":"Add upgrade-operator app to StarlingX","status":"MERGED","created":"2026-06-26 00:56:18.000000000","updated":"2026-06-29 23:36:00.000000000","submitted":"2026-06-29 20:35:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"994994-new-project","meta_rev_id":"1b3572e4d77d5c1bff6ddc3e30bea7ea28b10444","_number":994994,"virtual_id_number":994994,"owner":{"_account_id":32852,"name":"Enzo Candotti","email":"enzo.candotti@windriver.com","username":"ecandotti"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-29 20:35:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"recommended":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"all":[{"value":1,"date":"2026-06-29 18:23:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-06-26 16:17:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-06-29 20:25:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":26026,"name":"Greg Waines","email":"greg.waines@windriver.com","username":"gwaines"}],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26026,"name":"Greg Waines","email":"greg.waines@windriver.com","username":"gwaines"},{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-26 00:57:48.000000000","updated_by":{"_account_id":32852,"name":"Enzo Candotti","email":"enzo.candotti@windriver.com","username":"ecandotti"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-06-26 00:58:02.000000000","updated_by":{"_account_id":32852,"name":"Enzo Candotti","email":"enzo.candotti@windriver.com","username":"ecandotti"},"reviewer":{"_account_id":26026,"name":"Greg Waines","email":"greg.waines@windriver.com","username":"gwaines"},"state":"REVIEWER"},{"updated":"2026-06-26 01:06:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-29 04:06:27.000000000","updated_by":{"_account_id":32852,"name":"Enzo Candotti","email":"enzo.candotti@windriver.com","username":"ecandotti"},"reviewer":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"state":"REVIEWER"}],"messages":[{"id":"b575f570e35af1648b6bb2b5d1fd0bfb85dac96a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32852,"name":"Enzo Candotti","email":"enzo.candotti@windriver.com","username":"ecandotti"},"date":"2026-06-26 00:56:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7fcce59c3700d309eb5cfde25c095da243a7246","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32852,"name":"Enzo Candotti","email":"enzo.candotti@windriver.com","username":"ecandotti"},"date":"2026-06-26 00:56:57.000000000","message":"Topic set to new-project","accounts_in_message":[],"_revision_number":1},{"id":"f6665faad621537f57df4cc28508afd95737a409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 01:06:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af477e1be85e4a0c86608e0cace1f415\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4a91cde8f5e344b19c1a7546ed2f2aa7 : SUCCESS in 4m 59s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/80d7c22bf6b442a0ac337e512ffa3db6 : FAILURE in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"36886e6f53a664cb906130e1b75698b050c14798","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32852,"name":"Enzo Candotti","email":"enzo.candotti@windriver.com","username":"ecandotti"},"date":"2026-06-26 03:17:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"22fea04fb6b53fcfb297e394d8b78e84ad1a55de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-26 03:27:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d27b7205ac1d4ef6a6407c7072f31f64\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/5ca82bdf39e9417f813ddb25cfec4f06 : SUCCESS in 4m 57s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/09bc84249944456e9a3dd9c1caf5e054 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":2},{"id":"485139137e1ef830f5ca007cd5d170701ec42660","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-26 16:17:37.000000000","message":"Patch Set 2: Code-Review+2\n\nHolding approval for confirmation from Scott.","accounts_in_message":[],"_revision_number":2},{"id":"11f2d1bf611f953b610d8d6f4f285355dd53785a","author":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"},"date":"2026-06-29 18:23:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fe3277dc51e44c242b91a869c1dfc60b20c185f2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-29 20:25:15.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5c78e11781a02e6116c07244627424ab0dd83850","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 20:25:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f1e612ec22a28d0b8b75e602e230dc3811f80c48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 20:35:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c38d18b15b43a5b514a0bc52469059\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4976b3d7b3424679aa0d80c4dc546bb1 : SUCCESS in 4m 37s\n- project-config-gerrit https://zuul.opendev.org/t/openstack/build/dc649a37e4f94ef486c0a8b10bbecae7 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"97f72bba957f0b7f0ded2d4cb553e9c10db47b4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 20:35:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0aba51a2adc75b89e38332f9c157fb21be753fff","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-29 21:04:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcdf9caf4c1144719aa276bc25cd95f0\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/709f68f13bda46a7972c52f41b85f3c4 : SUCCESS in 28m 23s\n- infra-prod-manage-projects https://zuul.opendev.org/t/openstack/build/9f00769b56b14dcab8ae0829ed8df63a : SUCCESS in 20m 47s\n- infra-prod-service-zuul https://zuul.opendev.org/t/openstack/build/6b677111bd274d03b0ef237dc22e046e : SUCCESS in 5m 08s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/system-config/playbooks/zuul/run-production-bootstrap-bridge.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":2},{"id":"1b3572e4d77d5c1bff6ddc3e30bea7ea28b10444","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-06-29 23:36:00.000000000","message":"Patch Set 2:\n\nI\u0027ve added Scott as the initial member of the starlingx-app-upgrade-operator-core group.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ccc3d30d84f6de1eee963b2f70c1f2b4d2e12a8e","revisions":{"fe69f300ef1ef420ca088ff81ccb3faf3f58a1db":{"kind":"REWORK","_number":1,"created":"2026-06-26 00:56:18.000000000","uploader":{"_account_id":32852,"name":"Enzo Candotti","email":"enzo.candotti@windriver.com","username":"ecandotti"},"ref":"refs/changes/94/994994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/94/994994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/994994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/994994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/994994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/94/994994/1"}}},"commit":{"parents":[{"commit":"bd7657f12fe4f5e8f502edd1b277648fd70b7ee1","subject":"Merge \"Remove branch deletion powers from three new repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd7657f12fe4f5e8f502edd1b277648fd70b7ee1"}]}],"author":{"name":"Enzo Candotti","email":"Enzo.Candotti@windriver.com","date":"2026-06-26 00:54:14.000000000","tz":-180},"committer":{"name":"Enzo Candotti","email":"Enzo.Candotti@windriver.com","date":"2026-06-26 00:55:37.000000000","tz":-180},"subject":"Add upgrade-operator app to StarlingX","message":"Add upgrade-operator app to StarlingX\n\nIntroduce the Upgrade Operator FluxCD application\nto StarlingX.\n\nStory: 2011831\nTask: 54986\n\nChange-Id: Ic28d37f3056362697fcde97a4704a645a243581d\nSigned-off-by: Enzo Candotti \u003cEnzo.Candotti@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fe69f300ef1ef420ca088ff81ccb3faf3f58a1db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/fe69f300ef1ef420ca088ff81ccb3faf3f58a1db"}]},"branch":"refs/heads/master"},"ccc3d30d84f6de1eee963b2f70c1f2b4d2e12a8e":{"kind":"REWORK","_number":2,"created":"2026-06-26 03:17:02.000000000","uploader":{"_account_id":32852,"name":"Enzo Candotti","email":"enzo.candotti@windriver.com","username":"ecandotti"},"ref":"refs/changes/94/994994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/94/994994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/994994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/994994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/94/994994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/94/994994/2"}}},"commit":{"parents":[{"commit":"bd7657f12fe4f5e8f502edd1b277648fd70b7ee1","subject":"Merge \"Remove branch deletion powers from three new repo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/bd7657f12fe4f5e8f502edd1b277648fd70b7ee1"}]}],"author":{"name":"Enzo Candotti","email":"Enzo.Candotti@windriver.com","date":"2026-06-26 00:54:14.000000000","tz":-180},"committer":{"name":"Enzo Candotti","email":"Enzo.Candotti@windriver.com","date":"2026-06-26 03:16:54.000000000","tz":-180},"subject":"Add upgrade-operator app to StarlingX","message":"Add upgrade-operator app to StarlingX\n\nIntroduce the Upgrade Operator FluxCD application\nto StarlingX.\n\nStory: 2011831\nTask: 54986\n\nChange-Id: Ic28d37f3056362697fcde97a4704a645a243581d\nSigned-off-by: Enzo Candotti \u003cEnzo.Candotti@windriver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccc3d30d84f6de1eee963b2f70c1f2b4d2e12a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/ccc3d30d84f6de1eee963b2f70c1f2b4d2e12a8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28410,"name":"Scott Little","email":"scott.little@windriver.com","username":"slittle1"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
