)]}'
{"id":"openstack%2Fproject-config~998419","triplet_id":"openstack%2Fproject-config~master~Ic96860f3941593172d2c27dce47e1316cbc6a54c","project":"openstack/project-config","branch":"master","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2026-08-24 15:20:48.000000000","reason":"\u003cGERRIT_ACCOUNT_4146\u003e replied on the change","reason_account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-08-24 15:38:46.000000000","reason":"Change was submitted"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2026-08-22 14:02:29.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ic96860f3941593172d2c27dce47e1316cbc6a54c","subject":"Add check-sources propose job for kolla","status":"MERGED","created":"2026-07-23 08:44:15.000000000","updated":"2026-08-24 15:50:32.000000000","submitted":"2026-08-24 15:38:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"998419","meta_rev_id":"ece81f0bae690b182a9aa04784346be5f7a381d7","_number":998419,"virtual_id_number":998419,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-08-24 15:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2026-08-24 15:20:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2026-08-24 13:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2026-08-24 07:57:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-08-22 14:02:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":1,"date":"2026-08-24 15:20:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-07-23 08:56:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-07-23 09:23:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2026-08-21 17:35:51.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2026-08-21 17:57:03.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"CC"},{"updated":"2026-08-24 07:57:35.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2026-08-24 13:23:48.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2026-08-24 15:20:48.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"fe8bd31114caf51aa8bc594032acc05ca2cc9a18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 08:44:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f4bd659b019a028052703b33ba408846b3e45c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 08:56:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10d392bad6db4f40bbbcc67ccb3bfd0f\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/dd8f78c882754455a9879948b6e48ffd : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"fcd7ebc89ba0861a0dc11924509090160363cb42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-23 09:12:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"48807d8afaef9adaadf93cfeeb25546bfc281f09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-23 09:21:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35289715a6e0448f86b3b90e973d8a08\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/4f1caaf783a74854ba3d24270a4e5c6a : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"87154e0cd3fa34ff84d7c49a232a9ba48dc6033c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-13 07:01:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"1eceea4b1813a37d68307462d1b8a579294ddf8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-13 07:05:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5fb115d778fe8bd306eafd225078107ffefdeeba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 07:24:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e74a7254b2840769401aa174af96832\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/d95ac724600643ed958e1ea531ffe02c : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":4},{"id":"1c7fc3035a30b2b4bf7cff08ab31fcb32ffa0650","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-21 17:35:51.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d53a7a77de0cc7bfb81764353ed73c9b150e59d1","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-08-21 17:57:03.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f8132211fa3d7dadb5204ed69ec044054b5c4b3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-22 06:21:58.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"2f15848fd58bd5589da4c2a3034aae7c6f82b4e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-22 06:22:35.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"eb3849507425423b66806ec301277f52975dbac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 06:29:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a861979c74e746b2bcca6ad24be3e63e\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/cda1e14e44314be8a9cb8de4d3e2668d : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":6},{"id":"7af34b0b41ca42ab0c289f71d18f73baf11c8d2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-22 06:52:40.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"16fa856530a5ed57279247e1e1db5f5aa9b2d309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 06:53:20.000000000","message":"Patch Set 7: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/c/openstack/kolla/+/1001915 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"7c15ce75f6ee825bf61f5134874207a387f745bc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-22 06:56:11.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9168cc2603dfd6d53def5764f506edd7ff1504fe","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-22 06:56:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3476d66b3a6a6ae1f2346a29a4d0d03497a390be","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-22 06:56:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"036106a51401dfaec44ca52569d83b65d776eca8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-08-22 06:57:08.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":8},{"id":"92fadd393563ef35a0d8848329f3dc3ec507d048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-22 07:06:31.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdd03fc9c0e14e53b4b84e563ccdc51d\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/8e5effaf7d584d56b620b3f5c3d99b54 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":8},{"id":"e4f98289ccbcfd951340ca2d8e8d47e7d1a181e7","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2026-08-22 14:02:29.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"3a43ca653ff6e3da27ccdb274c5124e66caa3b93","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-08-24 07:57:35.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"5b7727172014919c358c9b943ce2047fcaf3eb07","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2026-08-24 13:23:48.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"beb8a7cd355f80065d665fcadd826c2e64117deb","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2026-08-24 15:20:48.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"b53bd92889f9228f1c737873d9bf0a9393650d3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 15:32:04.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"2f88e3e58c64e6734ba62a0b3485c226998ad4da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 15:38:46.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78e369edd0b84b8eb6264b3e20b8ed2b\n\n- openstack-zuul-jobs-linters https://zuul.opendev.org/t/openstack/build/a3b6ccf263b64269b184f3e202b90662 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":8},{"id":"e89a11a81f970861e4eb7f0f4190de9a642e417d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 15:38:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"ece81f0bae690b182a9aa04784346be5f7a381d7","tag":"autogenerated:zuul:deploy","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-24 15:50:32.000000000","message":"Patch Set 8:\n\nBuild succeeded (deploy pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f24b04a65424942a0e779ac5a94ed40\n\n- infra-prod-bootstrap-bridge https://zuul.opendev.org/t/openstack/build/6ad15678215f4c3486f2bee0910d5c1b : SUCCESS in 3m 09s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/system-config/playbooks/zuul/run-production-bootstrap-bridge.yaml for job infra-prod-bootstrap-bridge.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job infra-prod-bootstrap-bridge.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job infra-prod-bootstrap-bridge.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job infra-prod-bootstrap-bridge.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"982732aeb3b1cfe7ac28b04ec37d0274832efc5c","revisions":{"4b8b8f5210ded1c83984d5b64a0783742e7f8fff":{"kind":"REWORK","_number":1,"created":"2026-07-23 08:44:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/998419/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/998419/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/998419/1"}}},"commit":{"parents":[{"commit":"a85a4a55a45ae6616af60ddf08ffdf48e72c13ef","subject":"toggleWipState perm for Starlingx core groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a85a4a55a45ae6616af60ddf08ffdf48e72c13ef"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 08:41:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 08:44:13.000000000","tz":120},"subject":"Add check-sources propose job for kolla","message":"Add check-sources propose job for kolla\n\nKolla has developed a tool for updating sources.py, which store\nversions of dependencies like etcd, prometheus, etc.\n\nThis is a proposal to run it periodically to update them, instead\nof doing that manually.\n\nChange-Id: Ic96860f3941593172d2c27dce47e1316cbc6a54c\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b8b8f5210ded1c83984d5b64a0783742e7f8fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/4b8b8f5210ded1c83984d5b64a0783742e7f8fff"}]},"branch":"refs/heads/master"},"2fa42f9168b435a52099b1d0fcde83b4c3e5b782":{"kind":"REWORK","_number":2,"created":"2026-07-23 09:12:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/998419/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/998419/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/998419/2"}}},"commit":{"parents":[{"commit":"a85a4a55a45ae6616af60ddf08ffdf48e72c13ef","subject":"toggleWipState perm for Starlingx core groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a85a4a55a45ae6616af60ddf08ffdf48e72c13ef"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 08:41:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 09:11:50.000000000","tz":120},"subject":"Add check-sources propose job for kolla","message":"Add check-sources propose job for kolla\n\nKolla has developed a tool for updating sources.py, which store\nversions of dependencies like etcd, prometheus, etc.\n\nThis is a proposal to run it periodically to update them, instead\nof doing that manually.\n\nDefine a new job, because there\u0027s already a propose-updates job\nin openstack/kolla periodic config.\n\nChange-Id: Ic96860f3941593172d2c27dce47e1316cbc6a54c\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2fa42f9168b435a52099b1d0fcde83b4c3e5b782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/2fa42f9168b435a52099b1d0fcde83b4c3e5b782"}]},"branch":"refs/heads/master"},"7a0457dad5128bdf3486d9d81de3a127a80e533c":{"kind":"REWORK","_number":3,"created":"2026-08-13 07:01:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/998419/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/998419/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/998419/3"}}},"commit":{"parents":[{"commit":"a85a4a55a45ae6616af60ddf08ffdf48e72c13ef","subject":"toggleWipState perm for Starlingx core groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a85a4a55a45ae6616af60ddf08ffdf48e72c13ef"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 08:41:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-13 07:01:55.000000000","tz":120},"subject":"Add check-sources propose job for kolla","message":"Add check-sources propose job for kolla\n\nKolla has developed a tool for updating sources.py, which store\nversions of dependencies like etcd, prometheus, etc.\n\nThis is a proposal to run it periodically to update them, instead\nof doing that manually.\n\nDefine a new job, because there\u0027s already a propose-updates job\nin openstack/kolla periodic config and vendor check-sources.py\nwith identical content as in [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/1000800\n\nChange-Id: Ic96860f3941593172d2c27dce47e1316cbc6a54c\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7a0457dad5128bdf3486d9d81de3a127a80e533c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/7a0457dad5128bdf3486d9d81de3a127a80e533c"}]},"branch":"refs/heads/master"},"cd5961c1aff5e4ba5b3eb7afa00de77dbc5590b9":{"kind":"REWORK","_number":4,"created":"2026-08-13 07:05:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/998419/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/998419/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/998419/4"}}},"commit":{"parents":[{"commit":"a85a4a55a45ae6616af60ddf08ffdf48e72c13ef","subject":"toggleWipState perm for Starlingx core groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a85a4a55a45ae6616af60ddf08ffdf48e72c13ef"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 08:41:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-13 07:05:20.000000000","tz":120},"subject":"Add check-sources propose job for kolla","message":"Add check-sources propose job for kolla\n\nKolla has developed a tool for updating sources.py, which store\nversions of dependencies like etcd, prometheus, etc.\n\nThis is a proposal to run it periodically to update them, instead\nof doing that manually.\n\nDefine a new job, because there\u0027s already a propose-updates job\nin openstack/kolla periodic config and vendor check-sources.py\nwith identical content as in [1].\n\n[1]: https://review.opendev.org/c/openstack/kolla/+/1000800\n\nChange-Id: Ic96860f3941593172d2c27dce47e1316cbc6a54c\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cd5961c1aff5e4ba5b3eb7afa00de77dbc5590b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/cd5961c1aff5e4ba5b3eb7afa00de77dbc5590b9"}]},"branch":"refs/heads/master"},"b286a923d5bf805ed6c258fa4cd1963cbceed8ca":{"kind":"REWORK","_number":5,"created":"2026-08-22 06:21:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/998419/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/998419/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/998419/5"}}},"commit":{"parents":[{"commit":"a85a4a55a45ae6616af60ddf08ffdf48e72c13ef","subject":"toggleWipState perm for Starlingx core groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a85a4a55a45ae6616af60ddf08ffdf48e72c13ef"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 08:41:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-22 06:21:55.000000000","tz":120},"subject":"Add check-sources propose job for kolla","message":"Add check-sources propose job for kolla\n\nKolla has developed a tool for updating sources.py, which store\nversions of dependencies like etcd, prometheus, etc.\n\nThis is a proposal to run it periodically to update them, instead\nof doing that manually.\n\nDefine a new job, because there\u0027s already a propose-updates job\nin openstack/kolla periodic config and vendor check-sources.py\nwith identical content as in [1].\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/1001915\n\nChange-Id: Ic96860f3941593172d2c27dce47e1316cbc6a54c\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b286a923d5bf805ed6c258fa4cd1963cbceed8ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/b286a923d5bf805ed6c258fa4cd1963cbceed8ca"}]},"branch":"refs/heads/master"},"6ef338c0aa78487253d3ccfcb6d156d616fac772":{"kind":"REWORK","_number":6,"created":"2026-08-22 06:22:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/998419/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/998419/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/998419/6"}}},"commit":{"parents":[{"commit":"a85a4a55a45ae6616af60ddf08ffdf48e72c13ef","subject":"toggleWipState perm for Starlingx core groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a85a4a55a45ae6616af60ddf08ffdf48e72c13ef"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 08:41:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-22 06:22:35.000000000","tz":0},"subject":"Add check-sources propose job for kolla","message":"Add check-sources propose job for kolla\n\nKolla has developed a tool for updating sources.py, which store\nversions of dependencies like etcd, prometheus, etc.\n\nThis is a proposal to run it periodically to update them, instead\nof doing that manually.\n\nDefine a new job, because there\u0027s already a propose-updates job\nin openstack/kolla periodic config and vendor check-sources.py\nwith identical content as in [1].\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/1001915\n\nChange-Id: Ic96860f3941593172d2c27dce47e1316cbc6a54c\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ef338c0aa78487253d3ccfcb6d156d616fac772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/6ef338c0aa78487253d3ccfcb6d156d616fac772"}]},"branch":"refs/heads/master"},"915a7d7ab1332ede5e2ec7d8c91955a2d2e3586a":{"kind":"REWORK","_number":7,"created":"2026-08-22 06:52:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/998419/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/998419/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/998419/7"}}},"commit":{"parents":[{"commit":"a85a4a55a45ae6616af60ddf08ffdf48e72c13ef","subject":"toggleWipState perm for Starlingx core groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a85a4a55a45ae6616af60ddf08ffdf48e72c13ef"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 08:41:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-22 06:52:33.000000000","tz":120},"subject":"Add check-sources propose job for kolla","message":"Add check-sources propose job for kolla\n\nKolla has developed a tool for updating sources.py, which store\nversions of dependencies like etcd, prometheus, etc.\n\nThis is a proposal to run it periodically to update them, instead\nof doing that manually.\n\nDefine a new job, because there\u0027s already a propose-updates job\nin openstack/kolla periodic config and vendor check-sources.py\nwith identical content as in [1].\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/1001915\n\nChange-Id: Ic96860f3941593172d2c27dce47e1316cbc6a54c\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/915a7d7ab1332ede5e2ec7d8c91955a2d2e3586a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/915a7d7ab1332ede5e2ec7d8c91955a2d2e3586a"}]},"branch":"refs/heads/master"},"982732aeb3b1cfe7ac28b04ec37d0274832efc5c":{"kind":"REWORK","_number":8,"created":"2026-08-22 06:57:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/19/998419/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-config","ref":"refs/changes/19/998419/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-config refs/changes/19/998419/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-config refs/changes/19/998419/8"}}},"commit":{"parents":[{"commit":"a85a4a55a45ae6616af60ddf08ffdf48e72c13ef","subject":"toggleWipState perm for Starlingx core groups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/a85a4a55a45ae6616af60ddf08ffdf48e72c13ef"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-07-23 08:41:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2026-08-22 06:57:08.000000000","tz":0},"subject":"Add check-sources propose job for kolla","message":"Add check-sources propose job for kolla\n\nKolla has developed a tool for updating sources.py, which store\nversions of dependencies like etcd, prometheus, etc.\n\nThis is a proposal to run it periodically to update them, instead\nof doing that manually.\n\nDefine a new job, because there\u0027s already a propose-updates job\nin openstack/kolla periodic config and vendor check-sources.py\nwith identical content as in [1].\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/1001915\n\nChange-Id: Ic96860f3941593172d2c27dce47e1316cbc6a54c\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/982732aeb3b1cfe7ac28b04ec37d0274832efc5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-config/commit/982732aeb3b1cfe7ac28b04ec37d0274832efc5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
