)]}'
{"id":"openstack%2Fproject-navigator-data~458291","triplet_id":"openstack%2Fproject-navigator-data~master~I7da13dd1ae4fe2410cf1ceeb5dd55045ce448d83","project":"openstack/project-navigator-data","branch":"master","topic":"trove-project-navigator-data","hashtags":[],"change_id":"I7da13dd1ae4fe2410cf1ceeb5dd55045ce448d83","subject":"Add trove version history","status":"MERGED","created":"2017-04-20 01:49:50.000000000","updated":"2017-04-20 17:41:01.000000000","submitted":"2017-04-20 17:41:01.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3dad56a6e4f00cd5fbf130da29b09e0b4fdd8aa0","_number":458291,"virtual_id_number":458291,"owner":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":2,"date":"2017-04-20 17:41:01.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":2,"date":"2017-04-20 17:36:30.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":1,"date":"2017-04-20 17:36:30.000000000","_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":3,"name":"Jenkins","username":"jenkins"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-04-20 17:36:30.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2017-04-20 17:41:01.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"b79978f21a9b366eb940a0949a71b73d8ddd7d54","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-04-20 01:49:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31572bd1348dad9594e70cfacb0aa3e071eafd91","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-20 01:52:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-navigator-data-pep8-ubuntu-xenial http://logs.openstack.org/91/458291/1/check/gate-project-navigator-data-pep8-ubuntu-xenial/4c3c5b6/ : SUCCESS in 1m 34s\n- gate-project-navigator-data-tox-validate-ubuntu-xenial http://logs.openstack.org/91/458291/1/check/gate-project-navigator-data-tox-validate-ubuntu-xenial/3c4bcc2/ : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8d2bb55d8840b0d927fa5aeb7d25b222dc343570","author":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"date":"2017-04-20 02:03:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4fbc086edcb7c5105535732f8b7a89f755723943","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-20 02:06:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-navigator-data-pep8-ubuntu-xenial http://logs.openstack.org/91/458291/2/check/gate-project-navigator-data-pep8-ubuntu-xenial/90f3455/ : SUCCESS in 3m 03s\n- gate-project-navigator-data-tox-validate-ubuntu-xenial http://logs.openstack.org/91/458291/2/check/gate-project-navigator-data-tox-validate-ubuntu-xenial/079def9/ : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2098833eefdd224f81d4948b6b852f72ac978b04","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2017-04-20 17:36:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2fbc0bfa1f47b7ececb6864f91a667bf4015cfd6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-20 17:36:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"e72d1118726630090f588fd87ffe75a025f724d4","date":"2017-04-20 17:41:01.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"42fcf9c97d437e566186d87bacc89497ee5b5a7d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-20 17:41:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-project-navigator-data-pep8-ubuntu-xenial http://logs.openstack.org/91/458291/2/gate/gate-project-navigator-data-pep8-ubuntu-xenial/43d0ba7/ : SUCCESS in 1m 46s\n- gate-project-navigator-data-tox-validate-ubuntu-xenial http://logs.openstack.org/91/458291/2/gate/gate-project-navigator-data-tox-validate-ubuntu-xenial/0778ed8/ : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"414602dea2e2f49a88a450a63d7dca10588b3a91","revisions":{"11bf4a9fee66a32632f5f1777965bb359b7cdcda":{"kind":"REWORK","_number":1,"created":"2017-04-20 01:49:50.000000000","uploader":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"ref":"refs/changes/91/458291/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-navigator-data","ref":"refs/changes/91/458291/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/91/458291/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/91/458291/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/91/458291/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-navigator-data refs/changes/91/458291/1"}}},"commit":{"parents":[{"commit":"d1b1e9fd6e128af6bd6ceed70ef1c33059b9cf53","subject":"Merge \"Added freezer version data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/d1b1e9fd6e128af6bd6ceed70ef1c33059b9cf53"}]}],"author":{"name":"Amrith Kumar","email":"amrith@amrith.org","date":"2017-04-20 01:49:04.000000000","tz":-240},"committer":{"name":"Amrith Kumar","email":"amrith@amrith.org","date":"2017-04-20 01:49:47.000000000","tz":-240},"subject":"Add trove version history","message":"Add trove version history\n\nAdds the version histories for Trove from icehouse (trove was\nintegrated) through ocata.\n\nChange-Id: I7da13dd1ae4fe2410cf1ceeb5dd55045ce448d83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/11bf4a9fee66a32632f5f1777965bb359b7cdcda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/11bf4a9fee66a32632f5f1777965bb359b7cdcda"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d1b1e9fd6e128af6bd6ceed70ef1c33059b9cf53","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"414602dea2e2f49a88a450a63d7dca10588b3a91":{"kind":"REWORK","_number":2,"created":"2017-04-20 02:03:14.000000000","uploader":{"_account_id":9664,"name":"amrith","email":"amrith.kumar@gmail.com","username":"amrith"},"ref":"refs/changes/91/458291/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-navigator-data","ref":"refs/changes/91/458291/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/91/458291/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/91/458291/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/91/458291/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-navigator-data refs/changes/91/458291/2"}}},"commit":{"parents":[{"commit":"d1b1e9fd6e128af6bd6ceed70ef1c33059b9cf53","subject":"Merge \"Added freezer version data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/d1b1e9fd6e128af6bd6ceed70ef1c33059b9cf53"}]}],"author":{"name":"Amrith Kumar","email":"amrith@amrith.org","date":"2017-04-20 01:49:04.000000000","tz":-240},"committer":{"name":"Amrith Kumar","email":"amrith@amrith.org","date":"2017-04-20 02:03:05.000000000","tz":-240},"subject":"Add trove version history","message":"Add trove version history\n\nAdds the version histories for Trove from icehouse (trove was\nintegrated) through ocata.\n\nChange-Id: I7da13dd1ae4fe2410cf1ceeb5dd55045ce448d83\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/414602dea2e2f49a88a450a63d7dca10588b3a91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/414602dea2e2f49a88a450a63d7dca10588b3a91"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"d1b1e9fd6e128af6bd6ceed70ef1c33059b9cf53","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
