)]}'
{"id":"openstack%2Fproject-navigator-data~473072","triplet_id":"openstack%2Fproject-navigator-data~master~I1d09f5037d368080281d51a6d9c3e7f82b211085","project":"openstack/project-navigator-data","branch":"master","topic":"vitrage","hashtags":[],"change_id":"I1d09f5037d368080281d51a6d9c3e7f82b211085","subject":"Add the api versions for Vitrage.","status":"ABANDONED","created":"2017-06-11 10:46:17.000000000","updated":"2018-06-14 07:38:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cdc5b60841977cb9b07626738e24b98f5d1145f8","_number":473072,"virtual_id_number":473072,"owner":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2017-06-11 10:52:20.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2017-06-26 16:49:26.000000000","_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2017-06-13 12:26:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"value":1,"date":"2018-03-04 07:27:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-06-11 10:52:20.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-06-13 12:26:31.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2018-03-04 07:27:24.000000000","updated_by":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"reviewer":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"state":"REVIEWER"}],"messages":[{"id":"904b796606066700ae44509e3afc849ee24cd176","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2017-06-11 10:46:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8621df3df1be93a9dccda041fa39becd68f80f4","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2017-06-11 10:49:13.000000000","message":"Patch Set 2: Published edit on patch set 1","accounts_in_message":[],"_revision_number":2},{"id":"e3a0ac001e1f2760ff9bdbaf4d38ef4d070bfe79","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-11 10:52:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-project-navigator-data-pep8-ubuntu-xenial http://logs.openstack.org/72/473072/2/check/gate-project-navigator-data-pep8-ubuntu-xenial/1a9e5ae/ : SUCCESS in 2m 56s\n- gate-project-navigator-data-tox-validate-ubuntu-xenial http://logs.openstack.org/72/473072/2/check/gate-project-navigator-data-tox-validate-ubuntu-xenial/94a9e20/ : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":2},{"id":"63de98b04fbff21a476e91a9d686b080b06db254","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-06-13 12:26:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e291bdb753c8f16d674b68e3c01d74b14b6ea31c","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2017-06-26 16:49:26.000000000","message":"Patch Set 2:\n\nHi, I\u0027ll be happy to get another review on that. Thanks","accounts_in_message":[],"_revision_number":2},{"id":"a7ee905e4b65703ff936816866f50f00022c8802","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2018-03-04 07:27:24.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d77fa9d3dd3bd669099b61b6f0a94ef72faa7653","author":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"date":"2018-06-14 07:38:15.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef71ade17227ef7ddfd321bea5ac24b8b4925f60","revisions":{"b39f867aab80fb8e51ff47ea9a86728d19a41775":{"kind":"REWORK","_number":1,"created":"2017-06-11 10:46:17.000000000","uploader":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"ref":"refs/changes/72/473072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-navigator-data","ref":"refs/changes/72/473072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/72/473072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/72/473072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/72/473072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-navigator-data refs/changes/72/473072/1"}}},"commit":{"parents":[{"commit":"e58a41faaee17af89108af44165fba93b2edd667","subject":"Add Ocata release info for Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/e58a41faaee17af89108af44165fba93b2edd667"}]}],"author":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-06-11 09:56:05.000000000","tz":180},"committer":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-06-11 10:43:32.000000000","tz":180},"subject":"Add the api versions for Vitrage","message":"Add the api versions for Vitrage\n\nChange-Id: I1d09f5037d368080281d51a6d9c3e7f82b211085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/b39f867aab80fb8e51ff47ea9a86728d19a41775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/b39f867aab80fb8e51ff47ea9a86728d19a41775"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e58a41faaee17af89108af44165fba93b2edd667","is_merged_in_target_branch":true,"change_id":"I34cb512866c6890d54f1c72dd212cb290d230fda","change_number":460319,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"ef71ade17227ef7ddfd321bea5ac24b8b4925f60":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2017-06-11 10:49:13.000000000","uploader":{"_account_id":19159,"name":"Ifat Afek","email":"ifat.afek@nokia.com","username":"ifat_afek"},"ref":"refs/changes/72/473072/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-navigator-data","ref":"refs/changes/72/473072/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/72/473072/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/72/473072/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-navigator-data refs/changes/72/473072/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-navigator-data refs/changes/72/473072/2"}}},"commit":{"parents":[{"commit":"e58a41faaee17af89108af44165fba93b2edd667","subject":"Add Ocata release info for Manila","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/e58a41faaee17af89108af44165fba93b2edd667"}]}],"author":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-06-11 09:56:05.000000000","tz":180},"committer":{"name":"Ifat Afek","email":"ifat.afek@nokia.com","date":"2017-06-11 10:49:09.000000000","tz":0},"subject":"Add the api versions for Vitrage.","message":"Add the api versions for Vitrage.\n\nAlso, Vitrage is not a management tools - it is used to provide insights\nabout the topology and alarms in the cloud. Can you please move Vitrage\nunder the \"Data \u0026 Analytics\" category?\n\nChange-Id: I1d09f5037d368080281d51a6d9c3e7f82b211085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/ef71ade17227ef7ddfd321bea5ac24b8b4925f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-navigator-data/commit/ef71ade17227ef7ddfd321bea5ac24b8b4925f60"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"e58a41faaee17af89108af44165fba93b2edd667","is_merged_in_target_branch":true,"change_id":"I34cb512866c6890d54f1c72dd212cb290d230fda","change_number":460319,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
