)]}'
{"id":"openstack%2Fproject-team-guide~568994","triplet_id":"openstack%2Fproject-team-guide~master~If13f945dd5dafdaaa411577aa8803a986d309825","project":"openstack/project-team-guide","branch":"master","topic":"constraints-diagram","hashtags":[],"change_id":"If13f945dd5dafdaaa411577aa8803a986d309825","subject":"Diagram explaining constraints layers","status":"MERGED","created":"2018-05-17 01:22:20.000000000","updated":"2019-12-11 10:16:37.000000000","submitted":"2019-12-11 10:14:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"568994-1576059294201-51380ceb","meta_rev_id":"682bd89abed514614dbb610cd703bcf4587a68db","_number":568994,"virtual_id_number":568994,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"date":"2019-11-14 10:14:45.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-11 10:14:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-11-15 18:23:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-12-11 10:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":1,"date":"2019-12-11 10:10:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-05-17 01:47:58.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2019-07-01 12:14:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-10-28 10:07:48.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-11-14 10:14:45.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-11-15 18:23:19.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-12-11 10:10:19.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-12-11 10:14:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"69340a66391811680167379216c11b00579b406d","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2018-05-17 01:22:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a8bdf93563b0300640081c77185fb7845f3e63a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-05-17 01:30:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/94/568994/1/check/build-openstack-sphinx-docs/3fca194/html/ : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"40bd4628643a2ea54cee5ca04c18cb9a9c20907f","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2018-05-17 01:47:58.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks Ian  that looks great!","accounts_in_message":[],"_revision_number":1},{"id":"6920af3eddd878bc3b68005825743568bd6b4f3f","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2018-05-17 13:22:53.000000000","message":"Patch Set 1:\n\nThis is really nice, thanks!\n\nIn the current diagram the global lower constraint value should actually point to 1.1.0, because nova does not work with 1.0.1.\n\nYou use pyfoo-lib as a generic name for a dependency. I wonder if it would be better to use generic names for the services, too, for the same reason. \"Service A\" and \"Service B\" or something.\n\nIt would be useful to highlight that 2.0.1 is a newer release than we are testing with. We can do that in text. In fact, maybe most of the text should move to just regular text around the image? That would make it easier to read, change, and expand.\n\nIn the global requirement box, maybe you could say pyfoo-lib!\u003d1.1.2 and then we could explain the exclusion? Maybe that\u0027s cluttering this diagram up too much, though.","accounts_in_message":[],"_revision_number":1},{"id":"957795f32ef8a0632370ad07019743bc79aec5a2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2018-05-17 13:50:35.000000000","message":"Patch Set 1:\n\nLoving it! +1 on Doug\u0027s suggestions","accounts_in_message":[],"_revision_number":1},{"id":"0fbf89e7ac382fe5759899d6f761e690ae87a0e6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-03-16 02:07:05.000000000","message":"Patch Set 1:\n\nThis has been sitting for awhile. Looks like Doug had some good suggestions, but I would need to recheck to see the output at this point. If there are changes desired I am reluctant to do that.","accounts_in_message":[],"_revision_number":1},{"id":"cd04911d24808378f67b67882878b531d08ab66c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-07-01 12:14:20.000000000","message":"Patch Set 1:\n\nLooking at this again it occurs to me that none of this can be translated, and since the text is in an image, a non-English reader would not be able to easily copy out text into some sort of translation tool either. I wonder if there is a different way we can display this without making the entire thing an image.\n\nI do like the visual representation though, so I hope we can get something like this included at some point.","accounts_in_message":[],"_revision_number":1},{"id":"df095a352690cc107d87eeee5f3c4b81fc802b6d","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-07-22 08:53:17.000000000","message":"Patch Set 1:\n\n\u003e Looking at this again it occurs to me that none of this can be\n \u003e translated, and since the text is in an image, a non-English reader\n \u003e would not be able to easily copy out text into some sort of\n \u003e translation tool either. I wonder if there is a different way we\n \u003e can display this without making the entire thing an image.\n \u003e \n \u003e I do like the visual representation though, so I hope we can get\n \u003e something like this included at some point.\n\nA few (poor) substitutes for images can be: a table, using something like ascii text to \"draw\" a flow chart (of sorts)...\n\nWe do want the barrier to entry to be open to all, but not everything can be translated regardless. A suggestion would be to include \"subtitles\" of sort underneath the image to explain what it is. That can be translated roughly and would work.","accounts_in_message":[],"_revision_number":1},{"id":"b61c954f8ab802b1617f1a523381627c35a78a62","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-10-28 10:07:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"db86778ad2c33890bf9ba15f8bd80b3ac1e9a28c","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-13 11:21:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"17f71ce998652ca43eba5fda2a29b2dafeb6b096","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 11:29:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b08c94bb3b42928af87475c6adc6c2 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":2},{"id":"b8e0c1aa6b80e171c94cf63109592f7682e0116f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-14 10:13:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"35458bbb6412b0f44d543868b622bae94ac5c135","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-14 10:14:45.000000000","message":"Patch Set 3:\n\nNew diagrams incorporating Doug\u0027s comments, to hopefully finally merge this.","accounts_in_message":[],"_revision_number":3},{"id":"390681d89d5ee04130dcd7c74c184f12dafa2df9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 10:17:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e67d7b042c3d4d03bdb33910cfe6f489 : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":3},{"id":"1fd2054e0d1e25e13af62f7e124f0b3046f3ff6c","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-11-15 18:23:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"703d938f31484bffc34e6590bd6c9324c51dc0c7","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-12-11 10:10:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"239f72205383d1816d0a3d5c5feee8fa319e749c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 10:10:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1dc81e076071020b76eb3727b60c73f6365d3165","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 10:14:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd44aa7452614a138f610337363da0e8 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"e2363bfd723e7e3d4d503e46cb66aecb5a5c4f93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 10:14:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6167bf6366e377ed9f815ec26b5220bb0ca8690f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 10:16:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/6f5a2e25974d4fdcb95c9df8880ce9f9 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6c43aab43dc2ec77d7f0bd81e6f53bda7b2d207f","revisions":{"ca521725bc315755fa3ce7022506b278e2836c62":{"kind":"REWORK","_number":1,"created":"2018-05-17 01:22:20.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/94/568994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/94/568994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/1"}}},"commit":{"parents":[{"commit":"34eda618ab58a03a00ca8f1c332b4b7b0793f2e3","subject":"add some process instructions for updating dependencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/34eda618ab58a03a00ca8f1c332b4b7b0793f2e3"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2018-05-17 01:21:31.000000000","tz":600},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2018-05-17 01:22:14.000000000","tz":600},"subject":"Diagram explaining constraints layers","message":"Diagram explaining constraints layers\n\nThis is an attempt at a diagram explaining the various constraints\nthat apply to a project.\n\nChange-Id: If13f945dd5dafdaaa411577aa8803a986d309825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ca521725bc315755fa3ce7022506b278e2836c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ca521725bc315755fa3ce7022506b278e2836c62"}]},"branch":"refs/heads/master"},"32ecee4c4a3a5d1d0ea9bad7ad464d0531bbac53":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-13 11:21:24.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/94/568994/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/94/568994/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/2"}}},"commit":{"parents":[{"commit":"c7f10fd7582c07c8c61f1b0146d0db90594cb906","subject":"Add Note for Conflicts section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/c7f10fd7582c07c8c61f1b0146d0db90594cb906"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2018-05-17 01:21:31.000000000","tz":600},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-11-13 11:21:24.000000000","tz":0},"subject":"Diagram explaining constraints layers","message":"Diagram explaining constraints layers\n\nThis is an attempt at a diagram explaining the various constraints\nthat apply to a project.\n\nChange-Id: If13f945dd5dafdaaa411577aa8803a986d309825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/32ecee4c4a3a5d1d0ea9bad7ad464d0531bbac53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/32ecee4c4a3a5d1d0ea9bad7ad464d0531bbac53"}]},"branch":"refs/heads/master"},"6c43aab43dc2ec77d7f0bd81e6f53bda7b2d207f":{"kind":"REWORK","_number":3,"created":"2019-11-14 10:13:48.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/94/568994/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/94/568994/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/94/568994/3"}}},"commit":{"parents":[{"commit":"c7f10fd7582c07c8c61f1b0146d0db90594cb906","subject":"Add Note for Conflicts section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/c7f10fd7582c07c8c61f1b0146d0db90594cb906"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2018-05-17 01:21:31.000000000","tz":600},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-14 10:12:33.000000000","tz":60},"subject":"Diagram explaining constraints layers","message":"Diagram explaining constraints layers\n\nAdd a diagram to explain explaining the various constraints\nlayers that apply to a project dependencies.\n\nChange-Id: If13f945dd5dafdaaa411577aa8803a986d309825\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/6c43aab43dc2ec77d7f0bd81e6f53bda7b2d207f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/6c43aab43dc2ec77d7f0bd81e6f53bda7b2d207f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
