)]}'
{"id":"openstack%2Fproject-team-guide~665699","triplet_id":"openstack%2Fproject-team-guide~master~I5c5347a2354da80af8a8ec2530eedc79124d340b","project":"openstack/project-team-guide","branch":"master","topic":"documentation-update","hashtags":[],"change_id":"I5c5347a2354da80af8a8ec2530eedc79124d340b","subject":"Reviewing the PTL document for accessibility","status":"MERGED","created":"2019-06-17 14:58:38.000000000","updated":"2019-06-27 12:49:20.000000000","submitted":"2019-06-27 12:49:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":125,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665699-1561639760663-98f816f4","meta_rev_id":"2e754b7503168a2eb2d2ac4db5441109ddbf7ae7","_number":665699,"virtual_id_number":665699,"owner":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":2,"date":"2019-06-27 12:49:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"date":"2019-06-25 10:25:29.000000000","_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-06-25 11:51:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":2,"date":"2019-06-27 09:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-06-26 15:57:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":2,"date":"2019-06-27 09:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-06-27 12:44:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 16:05:13.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-06-18 13:36:07.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-06-21 12:22:45.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-06-21 15:12:55.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-06-21 16:21:20.000000000","updated_by":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"reviewer":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"state":"REVIEWER"},{"updated":"2019-06-24 04:11:41.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-06-26 15:57:20.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2019-06-27 09:03:40.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-06-27 09:32:35.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-27 12:44:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-06-27 12:49:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad98d9f5a04c5f7ac95efb4a041fef7703e9a5f8","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-17 14:58:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff3311bf7a8f37863d5cdbfdac0614998176ce70","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-17 15:00:38.000000000","message":"Topic set to documentation-update","accounts_in_message":[],"_revision_number":1},{"id":"60c21daad427b30f9b15ec4ba52f418d43e6b959","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 15:14:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/665699/1/check/openstack-tox-docs/9086727/html/ : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"dc2b93c9d1482110ce733caed3775835846fa9f1","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-17 15:49:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9423e8e1172a2e7296609ea0d537d03be5d399bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-17 16:02:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/665699/2/check/openstack-tox-docs/e65bcae/html/ : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7879f67ba9bffd6ac0607f42cb2a70df9be194ab","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-17 16:15:18.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)\n\nI only have minor suggestions. The updates look great. Thanks, Alex.","accounts_in_message":[],"_revision_number":2},{"id":"85d9e479447118e803e142026ba280e1ea5c8f75","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-18 05:16:30.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)\n\nThanks. almost lgtm. one suggestion on arranging the events related tasks under single heading.","accounts_in_message":[],"_revision_number":2},{"id":"f9db4fe2f11cf3b710dbe2f61e3f6e5b44eda10a","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-18 08:22:36.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3b574217df7a5e9868a4d8e698255e237cdd8651","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-18 09:10:06.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nWaiting for all the suggestions to be included before voting positively.","accounts_in_message":[],"_revision_number":2},{"id":"40fc92ed78e7eb7bd1d04a12e69e2f250b324dab","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-18 09:17:17.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"251d5bfed5a728ca49236615966f9cbba08b09a6","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-18 09:18:48.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"1236e136af503d10af02023746a74ce595ed95dc","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-18 09:19:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"47e155814439c845bc4ba1a28e1e15ec3a3c6fa9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:27:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/665699/4/check/openstack-tox-docs/76051f0/html/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":4},{"id":"9c8a036566263bb5fca7fbf4b427ba43cec2d589","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-18 09:29:40.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"b5d36860b50ecedba27c20c85750d3d4f9bbf968","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-18 09:35:34.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"8d4f6986436ea1b96cf0de746d367c070ef45906","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 09:38:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/665699/5/check/openstack-tox-docs/782ca3f/html/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":5},{"id":"675161938fbc7c2237a59919fc865627037ca6d7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-18 09:57:42.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8694b91608d20ef1534a6833856264d03894efd5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-18 10:08:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8b12fbc186e5554462ca0d1582b52ff29d48b45f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-18 10:08:56.000000000","message":"Patch Set 5: Code-Review+1\n\nthanks, lgtm.","accounts_in_message":[],"_revision_number":5},{"id":"3f3e8c62f490c5ae22a20f2084adce4a7fec7eea","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-18 10:28:56.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"ee2e61547ea9956f7d49cf513e45dc67394dffbf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 10:38:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/665699/6/check/openstack-tox-docs/58c2d0a/html/ : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":6},{"id":"69deec51268c22a802eadec482d6ee9bd30b7233","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-18 10:44:07.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"8ead1e04c80424514f24ad054399dc29748c9b96","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-18 11:32:37.000000000","message":"Patch Set 6:\n\n(3 comments)\n\nNice improvements, a few nits within","accounts_in_message":[],"_revision_number":6},{"id":"118f5dcbdb5de22f9dfb2ffb24ec8744db0ecab0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-18 11:43:59.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)\n\nLooks great! I agree with some of the things Chris pointed out, plus have a few more minor things, so going to vote -1 to make sure that\u0027s visible. Thanks for updating this!","accounts_in_message":[],"_revision_number":6},{"id":"1e01ecb320dcbc33a1358a70679ecad66e8fee86","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-18 12:28:42.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bb2ac54ffb95920408f0dd6fb14dae85aee2316d","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-18 13:20:00.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"cc06d1b9c7fea766adeaaae9ef957c0205bffd1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 13:25:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/665699/7/check/openstack-tox-docs/bc7876e/html/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":7},{"id":"3bdcba64fd1935996fdbed6675df7e98ad8937c4","author":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"date":"2019-06-18 13:36:07.000000000","message":"Patch Set 7:\n\n(9 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3ff6041e07ed18e8f4f7e2788a71f66514d6131f","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-18 15:53:09.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6553e3d48ed494a8af2996ee976f00874ee2df3d","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-18 15:54:08.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"20e8b91f10bdf63b92dbc67c7580d18c0c00f526","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-18 16:06:17.000000000","message":"Patch Set 7:\n\n(9 comments)","accounts_in_message":[],"_revision_number":7},{"id":"1ebb425d4955479c82ef990d6c534a408c533586","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-18 17:45:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f6b793b00124cee2e635917206b9d4231484fd61","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-18 17:59:56.000000000","message":"Patch Set 7:\n\n(13 comments)\n\nA bunch of comments inline, some I\u0027d like to see addressed before this merges, others that are just additional things which could be addressed in a followup. Which are which is left as an exercise for the reader. ;-)\n\nOr, ya know, ping me if you\u0027re not sure.","accounts_in_message":[],"_revision_number":7},{"id":"b963323ab25f4e8c65b095cb7951ad01df98f209","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-19 08:17:03.000000000","message":"Patch Set 7:\n\n(8 comments)","accounts_in_message":[],"_revision_number":7},{"id":"93f16f44c4fa57b76654d77043f9e0c6733e220d","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-19 09:15:19.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6a772f19a98bb4961636bf2ada96293fede0cc6c","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-19 09:15:26.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"0141841fc5ce34195c7495009ff42612fecb03ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-19 09:59:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/665699/8/check/openstack-tox-docs/d83ebab/html/ : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":8},{"id":"f2b3f75ef4e3d9ff407ac25b77bcc11346eafbeb","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-19 14:36:15.000000000","message":"Patch Set 8:\n\n(1 comment)\n\nwide ranging conment within about delegation\n\nI continue to think the changes thus far are worth having, even if we don\u0027t resolve the big issues. It\u0027s an improvement, and that\u0027s good.","accounts_in_message":[],"_revision_number":8},{"id":"87a2bd6c8099bcd455c79547afd00e758b471435","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-20 09:00:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bdb207f7f3dcd1fcb223f2920d05eb62c7a7f04a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-20 09:29:06.000000000","message":"Patch Set 8:\n\n(10 comments)","accounts_in_message":[],"_revision_number":8},{"id":"57b222cc5ea0841d78ce108f2dac67cbcdba038d","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-20 12:03:46.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2288b4a5f31d7c10d63c9e6db17fc5355fdb626b","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-21 09:26:34.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0fbe3b4239532d49c14ea29ba5f564c91bafa881","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-21 10:08:19.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"d70f0e26b02c98924fdabb114f03f1f910fb3a6e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 10:21:27.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/665699/9/check/openstack-tox-docs/0adaacf/html/ : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":9},{"id":"544bd2d0666815e3997160fbdf20b9b6937783cc","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-21 12:22:45.000000000","message":"Patch Set 9: Code-Review+1\n\n(9 comments)\n\nThis is a great improvement. Everything I\u0027ve noted within is a nit or just something to think about.","accounts_in_message":[],"_revision_number":9},{"id":"b740163bd433af4de7688d001ecea3b70cd1a590","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-21 13:49:12.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c241f1524bc05e9b6d64b14f075d982af24539fd","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-06-21 15:06:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1443f498b5a223995735f3ebf9ec50fabddeac23","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-06-21 15:12:55.000000000","message":"Patch Set 9: Code-Review+1\n\n(3 comments)\n\nI\u0027d be happy to have this merge now. There are some things we could tweak and improve, but I don\u0027t think there\u0027s anything I fundamentally disagree with at this point.","accounts_in_message":[],"_revision_number":9},{"id":"8f222861357c55c583b13e0e0de41490b98d6341","author":{"_account_id":8482,"name":"Colleen Murphy","email":"colleen@gazlene.net","username":"krinkle"},"date":"2019-06-21 16:21:20.000000000","message":"Patch Set 9: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"2da351e431d4a645c5d569fb64afe5d8832c7bb8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-06-24 04:11:41.000000000","message":"Patch Set 9: Code-Review+2\n\n(2 comments)\n\nlgtm. I am happy to add the links and other nits in a followup patch.","accounts_in_message":[],"_revision_number":9},{"id":"2dbc2247f976ed2d63cdb42b44cbaa7afdc4dbb2","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-24 12:20:45.000000000","message":"Patch Set 9:\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"ee7aae9bde9659bd2c9a04ad82abf32ab546a606","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-24 12:20:51.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"3da88cb9cabb46294a4feb059afa8051f1cedc02","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-24 12:21:26.000000000","message":"Patch Set 10:\n\nI\u0027d prefer to get this all done in this patch - if it\u0027s all the same. I don\u0027t want certain things to be forgotten (as they inevitably are) especially considering I will be away for 3 weeks and completely offline.","accounts_in_message":[],"_revision_number":10},{"id":"d7cca8607f8c7febfe3e1cd5a78127dcd434607a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 12:30:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/665699/10/check/openstack-tox-docs/862abeb/html/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":10},{"id":"c5544cbf8a6758af7902c8a436efc73774457f77","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-24 14:46:31.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"853979c8a5a5f728d78562f8849485a922cad513","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-24 14:51:59.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"42d4957b910f4c5596620a5cbb6674fa96967e7c","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-25 10:19:32.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"73d923ba0e2096f46106cfe34f5fa0495dabade6","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-25 10:19:40.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"6ea008007b60e3a02e14c8d8fa9d1a2608deb457","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-25 10:25:29.000000000","message":"Patch Set 11:\n\nUnless anyone has serious concerns with the content, I think this will be my last nit-related edit. If you have issues, you can propose a follow-up separately :)","accounts_in_message":[],"_revision_number":11},{"id":"28500a244bdf222303485549df5698e9615d20c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-25 10:32:45.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/665699/11/check/openstack-tox-docs/c04d7ae/html/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":11},{"id":"a928af1e7ca8d8da2ec2a6bfa7f2472bffab41e1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-25 11:50:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"04deb8a88f62694525798edab5968c901e64a28a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-25 11:51:12.000000000","message":"Patch Set 11: Code-Review+2\n\nThis looks good to me. Thanks!","accounts_in_message":[],"_revision_number":11},{"id":"af9bdd03db6a56e1229212408a0fb64dfb50f1d5","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2019-06-26 15:57:20.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"90b1450e64bb2f496714584eebda8e97b876532f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-27 09:03:40.000000000","message":"Patch Set 11: Code-Review+2\n\nGreat work!","accounts_in_message":[],"_revision_number":11},{"id":"8a5592fead6cf78db32f0c46cf530444a8461357","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-27 09:32:35.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"f20f9bbcca3d0c30a370a2b0adf7ada578d5ab13","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-27 12:44:41.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"06c020869688cd040e8bf101610f253fc314b2dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:44:53.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"1fd3ed18d557ffdfda0ab44ecbf4db87215e5897","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:49:20.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/99/665699/11/gate/openstack-tox-docs/8a9d6b2/html/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":11},{"id":"2e754b7503168a2eb2d2ac4db5441109ddbf7ae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 12:49:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"da3381f17636abfe53e7f35df4a503a4fae51ab5","revisions":{"117770d61e8e3e1d9d9a8f64b64776d3e30a81af":{"kind":"REWORK","_number":1,"created":"2019-06-17 14:58:38.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/99/665699/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/665699/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/1"}}},"commit":{"parents":[{"commit":"42c2fff08f799276e6fdd62e606ae4377b767c37","subject":"Update docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/42c2fff08f799276e6fdd62e606ae4377b767c37"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:57:08.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:58:36.000000000","tz":60},"subject":"Reviewing the PTL document for accessibility","message":"Reviewing the PTL document for accessibility\n\nAs per the TC conversation at the PTG (backed off\nthe PTL tips and tricks session in DEN from bnemec),\nthis change also adds a tips and tricks section to help\nease any potential pain points of becoming a PTL for\nsuch a large project.\n\nSuggestions welcome!\n\nChange-Id: I5c5347a2354da80af8a8ec2530eedc79124d340b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/117770d61e8e3e1d9d9a8f64b64776d3e30a81af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/117770d61e8e3e1d9d9a8f64b64776d3e30a81af"}]},"branch":"refs/heads/master"},"036a7caf6885abe2fa4cb9419d0c33292bd7dddc":{"kind":"REWORK","_number":2,"created":"2019-06-17 15:49:30.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/99/665699/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/665699/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/2"}}},"commit":{"parents":[{"commit":"42c2fff08f799276e6fdd62e606ae4377b767c37","subject":"Update docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/42c2fff08f799276e6fdd62e606ae4377b767c37"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:57:08.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-06-17 15:49:27.000000000","tz":0},"subject":"Reviewing the PTL document for accessibility","message":"Reviewing the PTL document for accessibility\n\nAs per the TC conversation at the PTG (backed off\nthe PTL tips and tricks session in DEN from bnemec),\nthis change also adds a tips and tricks section to help\nease any potential pain points of becoming a PTL for\nsuch a large project.\n\nSuggestions welcome!\n\nChange-Id: I5c5347a2354da80af8a8ec2530eedc79124d340b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/036a7caf6885abe2fa4cb9419d0c33292bd7dddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/036a7caf6885abe2fa4cb9419d0c33292bd7dddc"}]},"branch":"refs/heads/master"},"4344d3d986e30d6d89b462d59e6e38f1d550ee49":{"kind":"REWORK","_number":3,"created":"2019-06-18 09:17:17.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/99/665699/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/665699/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/3"}}},"commit":{"parents":[{"commit":"42c2fff08f799276e6fdd62e606ae4377b767c37","subject":"Update docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/42c2fff08f799276e6fdd62e606ae4377b767c37"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:57:08.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-06-18 09:17:14.000000000","tz":0},"subject":"Reviewing the PTL document for accessibility","message":"Reviewing the PTL document for accessibility\n\nAs per the TC conversation at the PTG (backed off\nthe PTL tips and tricks session in DEN from bnemec),\nthis change also adds a tips and tricks section to help\nease any potential pain points of becoming a PTL for\nsuch a large project.\n\nSuggestions welcome!\n\nChange-Id: I5c5347a2354da80af8a8ec2530eedc79124d340b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/4344d3d986e30d6d89b462d59e6e38f1d550ee49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/4344d3d986e30d6d89b462d59e6e38f1d550ee49"}]},"branch":"refs/heads/master"},"4910a52299512bea92880733adcabc2d5fcd7ef5":{"kind":"REWORK","_number":4,"created":"2019-06-18 09:18:48.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/99/665699/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/665699/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/4"}}},"commit":{"parents":[{"commit":"42c2fff08f799276e6fdd62e606ae4377b767c37","subject":"Update docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/42c2fff08f799276e6fdd62e606ae4377b767c37"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:57:08.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-06-18 09:18:46.000000000","tz":0},"subject":"Reviewing the PTL document for accessibility","message":"Reviewing the PTL document for accessibility\n\nAs per the TC conversation at the PTG (backed off\nthe PTL tips and tricks session in DEN from bnemec),\nthis change also adds a tips and tricks section to help\nease any potential pain points of becoming a PTL for\nsuch a large project.\n\nSuggestions welcome!\n\nChange-Id: I5c5347a2354da80af8a8ec2530eedc79124d340b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/4910a52299512bea92880733adcabc2d5fcd7ef5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/4910a52299512bea92880733adcabc2d5fcd7ef5"}]},"branch":"refs/heads/master"},"755e9442e85eff2c581032aa11ecc4156d4d6d1e":{"kind":"REWORK","_number":5,"created":"2019-06-18 09:29:40.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/99/665699/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/665699/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/5"}}},"commit":{"parents":[{"commit":"42c2fff08f799276e6fdd62e606ae4377b767c37","subject":"Update docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/42c2fff08f799276e6fdd62e606ae4377b767c37"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:57:08.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-06-18 09:29:34.000000000","tz":0},"subject":"Reviewing the PTL document for accessibility","message":"Reviewing the PTL document for accessibility\n\nAs per the TC conversation at the PTG (backed off\nthe PTL tips and tricks session in DEN from bnemec),\nthis change also adds a tips and tricks section to help\nease any potential pain points of becoming a PTL for\nsuch a large project.\n\nSuggestions welcome!\n\nChange-Id: I5c5347a2354da80af8a8ec2530eedc79124d340b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/755e9442e85eff2c581032aa11ecc4156d4d6d1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/755e9442e85eff2c581032aa11ecc4156d4d6d1e"}]},"branch":"refs/heads/master"},"1e8969aea7bd980550842994ec4d59bd8ac1c220":{"kind":"REWORK","_number":6,"created":"2019-06-18 10:28:56.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/99/665699/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/665699/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/6"}}},"commit":{"parents":[{"commit":"42c2fff08f799276e6fdd62e606ae4377b767c37","subject":"Update docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/42c2fff08f799276e6fdd62e606ae4377b767c37"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:57:08.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-06-18 10:28:53.000000000","tz":0},"subject":"Reviewing the PTL document for accessibility","message":"Reviewing the PTL document for accessibility\n\nAs per the TC conversation at the PTG (backed off\nthe PTL tips and tricks session in DEN from bnemec),\nthis change also adds a tips and tricks section to help\nease any potential pain points of becoming a PTL for\nsuch a large project.\n\nSuggestions welcome!\n\nChange-Id: I5c5347a2354da80af8a8ec2530eedc79124d340b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/1e8969aea7bd980550842994ec4d59bd8ac1c220"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/1e8969aea7bd980550842994ec4d59bd8ac1c220"}]},"branch":"refs/heads/master"},"fd477f5460addce641a8d12a8e310a4cde1bdf1d":{"kind":"REWORK","_number":7,"created":"2019-06-18 13:20:00.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/99/665699/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/665699/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/7"}}},"commit":{"parents":[{"commit":"42c2fff08f799276e6fdd62e606ae4377b767c37","subject":"Update docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/42c2fff08f799276e6fdd62e606ae4377b767c37"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:57:08.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-06-18 13:19:58.000000000","tz":0},"subject":"Reviewing the PTL document for accessibility","message":"Reviewing the PTL document for accessibility\n\nAs per the TC conversation at the PTG (backed off\nthe PTL tips and tricks session in DEN from bnemec),\nthis change also adds a tips and tricks section to help\nease any potential pain points of becoming a PTL for\nsuch a large project.\n\nSuggestions welcome!\n\nChange-Id: I5c5347a2354da80af8a8ec2530eedc79124d340b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/fd477f5460addce641a8d12a8e310a4cde1bdf1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/fd477f5460addce641a8d12a8e310a4cde1bdf1d"}]},"branch":"refs/heads/master"},"ef3ccb0236383272c74610a4a6ec993526ead938":{"kind":"REWORK","_number":8,"created":"2019-06-19 09:15:26.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/99/665699/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/665699/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/8"}}},"commit":{"parents":[{"commit":"42c2fff08f799276e6fdd62e606ae4377b767c37","subject":"Update docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/42c2fff08f799276e6fdd62e606ae4377b767c37"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:57:08.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-06-19 09:15:24.000000000","tz":0},"subject":"Reviewing the PTL document for accessibility","message":"Reviewing the PTL document for accessibility\n\nAs per the TC conversation at the PTG (backed off\nthe PTL tips and tricks session in DEN from bnemec),\nthis change also adds a tips and tricks section to help\nease any potential pain points of becoming a PTL for\nsuch a large project.\n\nSuggestions welcome!\n\nChange-Id: I5c5347a2354da80af8a8ec2530eedc79124d340b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ef3ccb0236383272c74610a4a6ec993526ead938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ef3ccb0236383272c74610a4a6ec993526ead938"}]},"branch":"refs/heads/master"},"509ef820847248fb5a1cf7c26ec3e63cba36e5bd":{"kind":"REWORK","_number":9,"created":"2019-06-21 10:08:19.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/99/665699/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/665699/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/9"}}},"commit":{"parents":[{"commit":"42c2fff08f799276e6fdd62e606ae4377b767c37","subject":"Update docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/42c2fff08f799276e6fdd62e606ae4377b767c37"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:57:08.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-06-21 10:08:16.000000000","tz":0},"subject":"Reviewing the PTL document for accessibility","message":"Reviewing the PTL document for accessibility\n\nAs per the TC conversation at the PTG (backed off\nthe PTL tips and tricks session in DEN from bnemec),\nthis change also adds a tips and tricks section to help\nease any potential pain points of becoming a PTL for\nsuch a large project.\n\nSuggestions welcome!\n\nChange-Id: I5c5347a2354da80af8a8ec2530eedc79124d340b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/509ef820847248fb5a1cf7c26ec3e63cba36e5bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/509ef820847248fb5a1cf7c26ec3e63cba36e5bd"}]},"branch":"refs/heads/master"},"40d48a10075489859adcfb933c27d2608e469df8":{"kind":"REWORK","_number":10,"created":"2019-06-24 12:20:51.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/99/665699/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/665699/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/10"}}},"commit":{"parents":[{"commit":"42c2fff08f799276e6fdd62e606ae4377b767c37","subject":"Update docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/42c2fff08f799276e6fdd62e606ae4377b767c37"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:57:08.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-06-24 12:20:41.000000000","tz":0},"subject":"Reviewing the PTL document for accessibility","message":"Reviewing the PTL document for accessibility\n\nAs per the TC conversation at the PTG (backed off\nthe PTL tips and tricks session in DEN from bnemec),\nthis change also adds a tips and tricks section to help\nease any potential pain points of becoming a PTL for\nsuch a large project.\n\nSuggestions welcome!\n\nChange-Id: I5c5347a2354da80af8a8ec2530eedc79124d340b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/40d48a10075489859adcfb933c27d2608e469df8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/40d48a10075489859adcfb933c27d2608e469df8"}]},"branch":"refs/heads/master"},"da3381f17636abfe53e7f35df4a503a4fae51ab5":{"kind":"REWORK","_number":11,"created":"2019-06-25 10:19:40.000000000","uploader":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"ref":"refs/changes/99/665699/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/665699/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/665699/11"}}},"commit":{"parents":[{"commit":"42c2fff08f799276e6fdd62e606ae4377b767c37","subject":"Update docs job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/42c2fff08f799276e6fdd62e606ae4377b767c37"}]}],"author":{"name":"Alexandra Settle","email":"asettle@suse.com","date":"2019-06-17 14:57:08.000000000","tz":60},"committer":{"name":"Alexandra Settle","email":"a.settle@outlook.com","date":"2019-06-25 10:19:14.000000000","tz":0},"subject":"Reviewing the PTL document for accessibility","message":"Reviewing the PTL document for accessibility\n\nAs per the TC conversation at the PTG (backed off\nthe PTL tips and tricks session in DEN from bnemec),\nthis change also adds a tips and tricks section to help\nease any potential pain points of becoming a PTL for\nsuch a large project.\n\nSuggestions welcome!\n\nChange-Id: I5c5347a2354da80af8a8ec2530eedc79124d340b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/da3381f17636abfe53e7f35df4a503a4fae51ab5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/da3381f17636abfe53e7f35df4a503a4fae51ab5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
