)]}'
{"id":"openstack%2Fproject-team-guide~666548","triplet_id":"openstack%2Fproject-team-guide~master~I4d86fe93f9701d9665c762630ae38881a119f721","project":"openstack/project-team-guide","branch":"master","topic":"cycle-automatic","hashtags":[],"change_id":"I4d86fe93f9701d9665c762630ae38881a119f721","subject":"Mention the new cycle-automatic release model","status":"MERGED","created":"2019-06-20 11:52:07.000000000","updated":"2019-07-18 14:18:15.000000000","submitted":"2019-07-18 14:18:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666548-1563459495329-31ef7618","meta_rev_id":"d97537e3fa08683ab1d1b55430438a0cc3d27d32","_number":666548,"virtual_id_number":666548,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-18 14:18:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-27 11:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-07-18 14:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-07-18 14:08:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-27 11:57:48.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-18 14:08:14.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-07-18 14:18:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47ce7a201b1ec46c75fbf7e829e31dad03406364","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-20 11:52:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"857261d3859c2f061028d1bc57f01f10cd1016d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 11:58:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/48/666548/1/check/openstack-tox-docs/b10fa08/html/ : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"39788afb9e9e87dc8e128be5f981f389b1dbd6bb","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-06-21 09:23:06.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c87e7ab774d2cb5fb5b1944c2bb4c5e032353296","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-27 09:43:55.000000000","message":"Patch Set 1: Code-Review+1\n\nClarifies the cycle-automatic. Nice. Could you please adapt with Alex comments?","accounts_in_message":[],"_revision_number":1},{"id":"c8acd3df050e3222e95f68afab210644a2555dca","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-27 09:52:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9002bafa754a7ecbb3874ee58be9956ee68cc353","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 09:58:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/48/666548/2/check/openstack-tox-docs/3a60dca/html/ : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":2},{"id":"1aae8c8cd746f6e4aa7f256c2b7bdc8352404692","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-27 11:57:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6970086b3572f208956a2e4c87fd1c2798815769","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-07-18 14:08:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3bd6e5fa3264a735807db21d589c68c06f435991","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-07-18 14:08:14.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"263c1353bcd26e425c4aae70479ec33bfecaf63a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 14:08:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1292b014435d8c6523b615254abdc5390ccffb0e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 14:18:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/48/666548/2/gate/openstack-tox-docs/d3bc2e4/html/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":2},{"id":"2850dfad76b00bfb4ccac35830869d9490ed0c45","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 14:18:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6e8124dedaa09b7f7c5e69a26a7bcf96feab4cf2","revisions":{"2cf0085997c949ed66e100f0542079d8548d5b85":{"kind":"REWORK","_number":1,"created":"2019-06-20 11:52:07.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/48/666548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/48/666548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/48/666548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/48/666548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/48/666548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/48/666548/1"}}},"commit":{"parents":[{"commit":"9e560ead2b5f9d5008f8fb11b7b2804c808c16dc","subject":"Merge \"Update Extended Maintenance transition description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/9e560ead2b5f9d5008f8fb11b7b2804c808c16dc"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-20 10:53:09.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-20 11:52:06.000000000","tz":120},"subject":"Mention the new cycle-automatic release model","message":"Mention the new cycle-automatic release model\n\nThe release team introduced a new release model in the Train\ncycle called cycle-automatic, to simplify handling of deliverables\nlike Tempest plugins. This should be reflected here.\n\nChange-Id: I4d86fe93f9701d9665c762630ae38881a119f721\nTask: 34188\nDepends-On: https://review.opendev.org/666546\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/2cf0085997c949ed66e100f0542079d8548d5b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/2cf0085997c949ed66e100f0542079d8548d5b85"}]},"branch":"refs/heads/master"},"6e8124dedaa09b7f7c5e69a26a7bcf96feab4cf2":{"kind":"REWORK","_number":2,"created":"2019-06-27 09:52:51.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/48/666548/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/48/666548/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/48/666548/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/48/666548/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/48/666548/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/48/666548/2"}}},"commit":{"parents":[{"commit":"9e560ead2b5f9d5008f8fb11b7b2804c808c16dc","subject":"Merge \"Update Extended Maintenance transition description\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/9e560ead2b5f9d5008f8fb11b7b2804c808c16dc"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-20 10:53:09.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-06-27 09:52:39.000000000","tz":120},"subject":"Mention the new cycle-automatic release model","message":"Mention the new cycle-automatic release model\n\nThe release team introduced a new release model in the Train\ncycle called cycle-automatic, to simplify handling of deliverables\nlike Tempest plugins. This should be reflected here.\n\nChange-Id: I4d86fe93f9701d9665c762630ae38881a119f721\nTask: 34188\nDepends-On: https://review.opendev.org/666546\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/6e8124dedaa09b7f7c5e69a26a7bcf96feab4cf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/6e8124dedaa09b7f7c5e69a26a7bcf96feab4cf2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
