)]}'
{"id":"openstack%2Fproject-team-guide~668027","triplet_id":"openstack%2Fproject-team-guide~master~I92e8ecd480937d55b968880c6c3d18885c3b12bf","project":"openstack/project-team-guide","branch":"master","topic":"doc8","hashtags":[],"change_id":"I92e8ecd480937d55b968880c6c3d18885c3b12bf","subject":"Enable doc8 linting","status":"MERGED","created":"2019-06-27 19:13:47.000000000","updated":"2019-07-01 11:33:40.000000000","submitted":"2019-07-01 11:33:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668027-1561980820319-4aedcafd","meta_rev_id":"c55f62387b388e6d622077eb54de709e09b57e49","_number":668027,"virtual_id_number":668027,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-07-01 11:33:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-06-28 13:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-01 11:29:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-01 11:29:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-28 13:44:14.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-07-01 11:29:38.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-07-01 11:33:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e016bb5d05afe8b4eaeb375c160dd8306a2c61e6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-27 19:13:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6f211670923887631a0f22d9e1ce94322f73ed2f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 19:21:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/27/668027/1/check/openstack-tox-docs/12a0b4d/ : POST_FAILURE in 2m 00s","accounts_in_message":[],"_revision_number":1},{"id":"274742a53d13114ca1cbf0c84e593d51c63b5254","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-06-27 19:22:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8fd9af744dc91c60dcacedee31b5fceb7727cc56","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 19:29:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/27/668027/2/check/openstack-tox-docs/680f523/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":2},{"id":"6d5c8680f5341be7c90a790b6558222f5f4f1f71","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-06-28 13:44:14.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"22c53f29d5e234ff409f2b52d4215c344d73b802","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2019-07-01 11:29:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\ntest all the things","accounts_in_message":[],"_revision_number":2},{"id":"dea5344b0e83e83c8fe4a1f74ac8b51c9ff1fb11","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 11:29:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8b6c14608a34ae779619d78128f4f3047d562135","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 11:33:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/27/668027/2/gate/openstack-tox-docs/8bf2dee/html/ : SUCCESS in 2m 35s","accounts_in_message":[],"_revision_number":2},{"id":"0932f4283d7ab93991fef05662e661ed736aafe5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 11:33:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef08132d5f1e37e17d38cb9fd0e708c6005fc153","revisions":{"d43dda041aa3f1c0568fc03b6c7142eb2a1bd80d":{"kind":"REWORK","_number":1,"created":"2019-06-27 19:13:47.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/27/668027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/27/668027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/668027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/668027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/668027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/27/668027/1"}}},"commit":{"parents":[{"commit":"ce15127abca68910c1cc4443b8f159ac38f5d764","subject":"Merge \"Reviewing the PTL document for accessibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ce15127abca68910c1cc4443b8f159ac38f5d764"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-06-27 19:12:49.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-06-27 19:12:49.000000000","tz":-300},"subject":"Enable doc8 linting","message":"Enable doc8 linting\n\nKeep our docs clean and try to avoid those cases where things are fine\nuntil a sphinx or other dependency gets updated and suddenly starts\nbreaking.\n\nChange-Id: I92e8ecd480937d55b968880c6c3d18885c3b12bf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/d43dda041aa3f1c0568fc03b6c7142eb2a1bd80d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/d43dda041aa3f1c0568fc03b6c7142eb2a1bd80d"}]},"branch":"refs/heads/master"},"ef08132d5f1e37e17d38cb9fd0e708c6005fc153":{"kind":"REWORK","_number":2,"created":"2019-06-27 19:22:01.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/27/668027/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/27/668027/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/668027/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/668027/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/668027/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/27/668027/2"}}},"commit":{"parents":[{"commit":"ce15127abca68910c1cc4443b8f159ac38f5d764","subject":"Merge \"Reviewing the PTL document for accessibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ce15127abca68910c1cc4443b8f159ac38f5d764"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-06-27 19:12:49.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-06-27 19:21:56.000000000","tz":-300},"subject":"Enable doc8 linting","message":"Enable doc8 linting\n\nKeep our docs clean and try to avoid those cases where things are fine\nuntil a sphinx or other dependency gets updated and suddenly starts\nbreaking.\n\nChange-Id: I92e8ecd480937d55b968880c6c3d18885c3b12bf\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ef08132d5f1e37e17d38cb9fd0e708c6005fc153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ef08132d5f1e37e17d38cb9fd0e708c6005fc153"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
