)]}'
{"id":"openstack%2Fproject-team-guide~682381","triplet_id":"openstack%2Fproject-team-guide~master~I92542012108f0aa07e28968479cdaddf7e06301d","project":"openstack/project-team-guide","branch":"master","topic":"em-eol","hashtags":[],"change_id":"I92542012108f0aa07e28968479cdaddf7e06301d","subject":"Elaborate on Extended Maintenance to EOL transition","status":"MERGED","created":"2019-09-16 14:57:49.000000000","updated":"2019-11-22 14:12:24.000000000","submitted":"2019-11-22 14:10:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682381-1574431848313-8eeca06d","meta_rev_id":"5a709b38a29980e260b264a9aa2997c933871d24","_number":682381,"virtual_id_number":682381,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-09-20 15:24:33.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-22 14:10:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-09-17 09:15:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-20 12:10:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-22 14:06:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-20 12:10:44.000000000","updated_by":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"reviewer":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"state":"REVIEWER"},{"updated":"2019-11-22 14:06:09.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-11-22 14:10:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8e7a881d419e265cdc322b8069d2be913b00735","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-16 14:57:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8526af40b338eca649f3056036d9627e08ba8faa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 15:02:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7f7aa0b6ec472ab6210ab636ea2263 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2226e88300c436d89df3af421efce8a1f4622618","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-09-17 09:15:55.000000000","message":"Patch Set 1: Code-Review+2\n\nWorks for me","accounts_in_message":[],"_revision_number":1},{"id":"fe3dc7dcf676737243ceecf296ce4ee85e7ffcaf","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-09-20 12:10:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7389846526f5890906e4dfbe65af73c6c1b71127","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-09-20 12:11:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"580964b13d2abbd9d5d76f0813819b304b6dff5f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-09-20 15:24:33.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n\nNot sure. Seems like there could be a lot of different reasons. Not really sure if we need to call them out.","accounts_in_message":[],"_revision_number":1},{"id":"0395409656a7411ab18cd25631497fc65e11cc6c","author":{"_account_id":10607,"name":"Alexandra Settle","email":"a.settle@outlook.com","username":"asettle"},"date":"2019-09-23 10:17:21.000000000","message":"Patch Set 1:\n\n\u003e \u003e (1 comment)\n \u003e \n \u003e Not sure. Seems like there could be a lot of different reasons. Not\n \u003e really sure if we need to call them out.\n\nExamples help. I mean, I guess my thinking is if we include something like this - anyone can create their own reasoning to end a branch life and assure you it\u0027s valid. If we provide some guidelines as to what this definition means, it\u0027s less likely to be taken for granted.","accounts_in_message":[],"_revision_number":1},{"id":"ad05ed6bb8959be15f935b2608c5da0d5f55d194","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-22 14:06:09.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"10328bac4291b10d92f05b956ba940046c7905a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 14:06:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1db58250ab9c6e431cd49b16d9bd3dabfe6e75fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 14:10:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b75c4f3a4da4fbea9a8da614eaceebd : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"457751d9f9f2ca3f945ca967d06885d97d8c0f1c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 14:10:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f85a8977f8c97840b828506ac8b1b3015ca68d5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-22 14:12:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/7331c26566034e3eb8b88e0ab16d3540 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f12005bb8b1169c14f72a14b1bc4c8fdce4885a","revisions":{"6f12005bb8b1169c14f72a14b1bc4c8fdce4885a":{"kind":"REWORK","_number":1,"created":"2019-09-16 14:57:49.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/81/682381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/81/682381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/81/682381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/81/682381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/81/682381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/81/682381/1"}}},"commit":{"parents":[{"commit":"fc75b0a74796f2f6ac4f253160b9975d4690b79b","subject":"Merge \"Allow multiple releases in cycle-automatic\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/fc75b0a74796f2f6ac4f253160b9975d4690b79b"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-16 14:54:54.000000000","tz":-300},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-09-16 14:57:36.000000000","tz":-300},"subject":"Elaborate on Extended Maintenance to EOL transition","message":"Elaborate on Extended Maintenance to EOL transition\n\nThis adds a little more wording around end of life for stable branches\nto make sure it\u0027s clear that a team can decide to go directly from EM to\nEOL without needing to wait 6 months for an \"Unmaintained\" phase to\ncomplete.\n\nAlso fixes some minor nits while I\u0027m here.\n\nChange-Id: I92542012108f0aa07e28968479cdaddf7e06301d\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/6f12005bb8b1169c14f72a14b1bc4c8fdce4885a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/6f12005bb8b1169c14f72a14b1bc4c8fdce4885a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
