)]}'
{"id":"openstack%2Fproject-team-guide~693829","triplet_id":"openstack%2Fproject-team-guide~master~I6a4efb1e703f6634cf2d260480153d0ceae3a46f","project":"openstack/project-team-guide","branch":"master","topic":"stable-conflicts","hashtags":[],"change_id":"I6a4efb1e703f6634cf2d260480153d0ceae3a46f","subject":"Add Note for Conflicts section","status":"MERGED","created":"2019-11-12 12:13:38.000000000","updated":"2019-11-12 13:48:15.000000000","submitted":"2019-11-12 13:47:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693829-1573566423530-2f2a50bb","meta_rev_id":"63080330695912fe8b7d1fcb63f6d3abfa32e170","_number":693829,"virtual_id_number":693829,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-12 13:47:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-11-12 13:13:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-12 13:41:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-12 13:41:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-12 13:13:43.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-12 13:41:24.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-11-12 13:47:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cff13765f47e0150a5c41e53bf6b3253941d7eea","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-11-12 12:13:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e9519ea281e8f3c534f30c1614afb62bc803c50","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-11-12 12:20:29.000000000","message":"Topic set to stable-conflicts","accounts_in_message":[],"_revision_number":1},{"id":"6319e976bac96fc616ef02abaa9e571604e67c03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 12:20:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e76302a62ae4fc48049cea63c06b623 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ce1e301091bd721252463dd8ead50e840b638f4b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-12 12:31:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41c38a1e6f37f111ebdbbba2cb36662e09c85696","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-11-12 12:50:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ee06754f768b9af287f969d832cb37179e6c55d9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-11-12 12:52:06.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks Sean! :) Patch updated.","accounts_in_message":[],"_revision_number":1},{"id":"a56e52c4a954b29b8060aa24ac04c7b9f4306b09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 12:55:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a5b1712ccc0446d8a582daf7ec50d75 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d11ba246d664aac68a4e91ee93eff91f5800cd71","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-12 13:13:43.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":2},{"id":"5eb21d2778ab7e8666dc55466961c0b6aa7cecd3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-12 13:41:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1acf81b1bc6b682e37c652194b21563d72b318ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 13:41:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e3d0ae0203e389a7273959a3b987dc367cd965e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 13:47:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c388ce182d7496280c33c908c4f5bc7 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f1947752ddf6086dd2deffc1b1f92ae9f56c367b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 13:47:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"63080330695912fe8b7d1fcb63f6d3abfa32e170","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 13:48:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/6430f27059b14842bc9b64baa896d5d9 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c7f10fd7582c07c8c61f1b0146d0db90594cb906","revisions":{"49466fcd864d8790f33fa220218a43aab9798ad5":{"kind":"REWORK","_number":1,"created":"2019-11-12 12:13:38.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/29/693829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/29/693829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/29/693829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/29/693829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/29/693829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/29/693829/1"}}},"commit":{"parents":[{"commit":"0e80c32f56a013b99fa9118a90277295e9fa2e23","subject":"Give an example of dealing with backport conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/0e80c32f56a013b99fa9118a90277295e9fa2e23"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-11-12 12:09:44.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-11-12 12:09:44.000000000","tz":60},"subject":"Add Note for Conflicts section","message":"Add Note for Conflicts section\n\nIn some cases the cherry-picked patch contains the original patch\u0027s\n\"Conflicts\" lines, that were valid only on the source branch (the\npatch was cherry-picked from). To help the reviewers these should\nbe removed from the commit message.\n\nChange-Id: I6a4efb1e703f6634cf2d260480153d0ceae3a46f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/49466fcd864d8790f33fa220218a43aab9798ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/49466fcd864d8790f33fa220218a43aab9798ad5"}]},"branch":"refs/heads/master"},"c7f10fd7582c07c8c61f1b0146d0db90594cb906":{"kind":"REWORK","_number":2,"created":"2019-11-12 12:50:34.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/29/693829/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/29/693829/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/29/693829/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/29/693829/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/29/693829/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/29/693829/2"}}},"commit":{"parents":[{"commit":"0e80c32f56a013b99fa9118a90277295e9fa2e23","subject":"Give an example of dealing with backport conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/0e80c32f56a013b99fa9118a90277295e9fa2e23"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-11-12 12:09:44.000000000","tz":60},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2019-11-12 12:49:34.000000000","tz":60},"subject":"Add Note for Conflicts section","message":"Add Note for Conflicts section\n\nIn some cases the cherry-picked patch contains the original patch\u0027s\n\"Conflicts\" lines, that were valid only on the source branch (the\npatch was cherry-picked from). To help the reviewers these should\nbe removed from the commit message.\n\nChange-Id: I6a4efb1e703f6634cf2d260480153d0ceae3a46f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/c7f10fd7582c07c8c61f1b0146d0db90594cb906"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/c7f10fd7582c07c8c61f1b0146d0db90594cb906"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
