)]}'
{"id":"openstack%2Fproject-team-guide~712534","triplet_id":"openstack%2Fproject-team-guide~master~I23870f8299dbc2c05568502e39c426de299db4e9","project":"openstack/project-team-guide","branch":"master","topic":"retire","hashtags":[],"change_id":"I23870f8299dbc2c05568502e39c426de299db4e9","subject":"Add some repository documentation","status":"MERGED","created":"2020-03-11 19:02:48.000000000","updated":"2020-03-11 20:14:37.000000000","submitted":"2020-03-11 20:12:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712534-1583957568896-4545ba28","meta_rev_id":"5904c92719ba9c5dc94335bdb952132ade4210ec","_number":712534,"virtual_id_number":712534,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-11 20:12:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-11 19:48:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-11 20:02:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-11 20:02:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-11 19:35:06.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2020-03-11 19:48:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-03-11 20:02:29.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-03-11 20:12:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"766ce0f239cc812cf65aba7f8bbff7d21f2e22e7","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-11 19:02:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae2c39243b38171cca59681d53f7d9d082ecb4d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 19:09:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925c4675dbd74b0cb350dc99b79efe74 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"5fd122cf83c7cd359b5595d50761e5daaffed0f4","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-11 19:14:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12ba8f73a4d1ac662868009369a805e9ddc355e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 19:27:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fc4f7a0213453dab7699df01ed3d13 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":2},{"id":"2fe2995019c2ea76de9f2b8bffec0b40b143affa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-11 19:28:54.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"409ace6c5c06d1f529963e62f2916dcad9c9e57d","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-11 19:30:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7cc402ba241f516ed7cdb8a43324d1621a380975","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-11 19:30:54.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d207ddb90331060dda920ee29d498fde3304cabd","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2020-03-11 19:35:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5d6d27b5261d2c3318aa0e9045d0ed5728f84574","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-03-11 19:35:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7f4b100890725dcb90f311bcc5f47788386ea72c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 19:41:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c4ec6aaaf0d4499aa03503efe8e8781 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":4},{"id":"c12d1a0b70e9bd807c2c3fc113af8149d82cf0db","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-11 19:48:15.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e9b16284e5519075bc861977c6ec9f378c62a705","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-03-11 20:02:29.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"36cfffa09b548ee2311cecd3caa64a54a5f1e933","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 20:02:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"013a6153dae2ed88dc093b1912b8ff72990e4ddb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 20:12:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c28719c3a55450a8fc7dea66ae6fadb : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":4},{"id":"61bf95d2de455da0b0f38623504078b87e88ccd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 20:12:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"0048c7b55e2ea87aa38eda541004cd9cf37d4468","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 20:14:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/d99facfd697f4bb094ba454eb9336a50 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f8e971d50cf3f0e453ff14e12934403474f40706","revisions":{"2ef68fd6f54d3ae01caadb01ac206aa3e383a5f1":{"kind":"REWORK","_number":1,"created":"2020-03-11 19:02:48.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/34/712534/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/34/712534/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/1"}}},"commit":{"parents":[{"commit":"a7f4b96ade3f7d470c9588d9597b6529463575c6","subject":"Merge \"Diagram explaining constraints layers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/a7f4b96ade3f7d470c9588d9597b6529463575c6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-11 19:01:25.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-11 19:02:40.000000000","tz":60},"subject":"Add new repository documentation","message":"Add new repository documentation\n\nChange https://review.opendev.org/#/c/711827 removes OpenStack specific\ncontent from the OpenDev Manual (previously Infra Manual), move\nsome OpenStack specific content to a new file in this repo.\n\nChange-Id: I23870f8299dbc2c05568502e39c426de299db4e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/2ef68fd6f54d3ae01caadb01ac206aa3e383a5f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/2ef68fd6f54d3ae01caadb01ac206aa3e383a5f1"}]},"branch":"refs/heads/master"},"7b18b115e47041584111f3f9173d612ac623f489":{"kind":"REWORK","_number":2,"created":"2020-03-11 19:14:57.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/34/712534/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/34/712534/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/2"}}},"commit":{"parents":[{"commit":"a7f4b96ade3f7d470c9588d9597b6529463575c6","subject":"Merge \"Diagram explaining constraints layers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/a7f4b96ade3f7d470c9588d9597b6529463575c6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-11 19:01:25.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-11 19:14:25.000000000","tz":60},"subject":"Add some repository documentation","message":"Add some repository documentation\n\nChange https://review.opendev.org/#/c/711827 removes OpenStack specific\ncontent from the OpenDev Manual (previously Infra Manual), move\nsome OpenStack specific content to a new file in this repo.\n\nThis change adds information about project renames and retirements.\n\nChange-Id: I23870f8299dbc2c05568502e39c426de299db4e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/7b18b115e47041584111f3f9173d612ac623f489"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/7b18b115e47041584111f3f9173d612ac623f489"}]},"branch":"refs/heads/master"},"5c5850a631d2a28f301eab478750b7708ff531c4":{"kind":"REWORK","_number":3,"created":"2020-03-11 19:30:44.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/34/712534/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/34/712534/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/3"}}},"commit":{"parents":[{"commit":"a7f4b96ade3f7d470c9588d9597b6529463575c6","subject":"Merge \"Diagram explaining constraints layers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/a7f4b96ade3f7d470c9588d9597b6529463575c6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-11 19:01:25.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-11 19:30:38.000000000","tz":60},"subject":"Add some repository documentation","message":"Add some repository documentation\n\nChange https://review.opendev.org/#/c/711827 removes OpenStack specific\ncontent from the OpenDev Manual (previously Infra Manual), move\nsome OpenStack specific content to a new file in this repo.\n\nThis change adds information about project renames and retirements.\n\nChange-Id: I23870f8299dbc2c05568502e39c426de299db4e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/5c5850a631d2a28f301eab478750b7708ff531c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/5c5850a631d2a28f301eab478750b7708ff531c4"}]},"branch":"refs/heads/master"},"f8e971d50cf3f0e453ff14e12934403474f40706":{"kind":"REWORK","_number":4,"created":"2020-03-11 19:35:55.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/34/712534/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/34/712534/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/34/712534/4"}}},"commit":{"parents":[{"commit":"a7f4b96ade3f7d470c9588d9597b6529463575c6","subject":"Merge \"Diagram explaining constraints layers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/a7f4b96ade3f7d470c9588d9597b6529463575c6"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-11 19:01:25.000000000","tz":60},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-03-11 19:35:34.000000000","tz":60},"subject":"Add some repository documentation","message":"Add some repository documentation\n\nChange https://review.opendev.org/#/c/711827 removes OpenStack specific\ncontent from the OpenDev Manual (previously Infra Manual), move\nsome OpenStack specific content to a new file in this repo.\n\nThis change adds information about project renames and retirements.\n\nChange-Id: I23870f8299dbc2c05568502e39c426de299db4e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/f8e971d50cf3f0e453ff14e12934403474f40706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/f8e971d50cf3f0e453ff14e12934403474f40706"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
