)]}'
{"id":"openstack%2Fproject-team-guide~740727","triplet_id":"openstack%2Fproject-team-guide~master~Ib33823e449ab98455f83a5be703b69ca3b886039","project":"openstack/project-team-guide","branch":"master","topic":"import-zuulv3-doc","hashtags":[],"change_id":"Ib33823e449ab98455f83a5be703b69ca3b886039","subject":"Import the Zuul v3 porting guide","status":"MERGED","created":"2020-07-13 16:28:53.000000000","updated":"2020-07-20 13:41:26.000000000","submitted":"2020-07-20 13:39:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740727-1595252340431-e29519fa","meta_rev_id":"67820cd4e49401ed44526263c453b8daa50e288f","_number":740727,"virtual_id_number":740727,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-07-20 12:19:37.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-20 13:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-07-14 13:13:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-20 13:34:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-20 13:34:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-14 13:13:25.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-07-20 13:34:00.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-07-20 13:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"afbf1333c7b9be803cf6f2a905151cd867bcf15a","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-13 16:28:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a809ded07ef7b6c12c8df43ee01bb3339ca0373","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-13 16:29:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"40bf543916b1abaa18ba23fc83b3d1c90d3bf7da","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-13 16:34:36.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nI wonder if some of the \"we\u0027re about to move off of Jenkins\" sections should just be removed now.","accounts_in_message":[],"_revision_number":2},{"id":"ccfb8ce03d60659949049439e073ca5bcd6d7eae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 16:40:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/229e51d5eb064408817469923dd87fee : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"2ce027ae8e93973d5c5bb82b58c6660af826bbe2","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-13 16:50:39.000000000","message":"Patch Set 2:\n\n(2 comments)\n\n\u003e (2 comments)\n \u003e \n \u003e I wonder if some of the \"we\u0027re about to move off of Jenkins\"\n \u003e sections should just be removed now.\n\nMaybe a follow-up change?","accounts_in_message":[],"_revision_number":2},{"id":"a346165750706ece0469cb74dd97b88084dcb034","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-13 17:25:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4ab89f0045523f75f20e0e2647acc7ae0ff84d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 17:33:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cdf460f10414c3ea7da9f1810956e3d : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":3},{"id":"cd54961e4518a5d5782d2003ea24606ebab44c5f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-13 18:21:56.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"c39a52145e0a0eac538620cb2591e73f3ea8ce9f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-14 01:37:29.000000000","message":"Patch Set 3: Code-Review-1\n\n(13 comments)\n\nthanks for doing this, few comments inline especially adding ref links as this guide is not in zuulv3 doc now and name without link can leads to blank concept for many of those.","accounts_in_message":[],"_revision_number":3},{"id":"bf1ecbdbe7b63205930c7ee4e710730c1a82f01e","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-14 08:44:14.000000000","message":"Patch Set 3:\n\nApart from maybe fixing the link to https://zuul-ci.org/docs/zuul/reference/job_def.html#attr-job.files, I think all the other changes are more fit for a follow-up. This is an import, the document should be almost the same (minus a few typographical issues). Improvements will happen later.","accounts_in_message":[],"_revision_number":3},{"id":"6e1e860424f37297ebe22f2f99e01474a1cbeab7","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-14 08:46:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad42c7e362fcffc8d1b0696494f2a19bf4f910f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 08:57:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61c2985b97b247d2bcc00d9ac97e46fa : FAILURE in 4m 32s","accounts_in_message":[],"_revision_number":4},{"id":"3a7d886790a6d433352d1724705d7ca8579e56ff","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-14 11:26:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6a226a5cc873772ffbe36b56a9c18957d1a75d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 11:32:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/638164e22e594bfd96b4e09a63400301 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":5},{"id":"b55a0a90d702eb4911bd75b8c10a3973f4b5bba8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-07-14 13:13:25.000000000","message":"Patch Set 5: Code-Review+2\n\n\u003e Apart from maybe fixing the link to https://zuul-ci.org/docs/zuul/reference/job_def.html#attr-job.files,\n \u003e I think all the other changes are more fit for a follow-up. This is\n \u003e an import, the document should be almost the same (minus a few\n \u003e typographical issues). Improvements will happen later.\n\n++","accounts_in_message":[],"_revision_number":5},{"id":"7f278bad1761bb4218f835b4035119b47c82997f","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-20 10:40:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7766dfce20a3b7cb4775c162b9056de25d6741bd","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-07-20 12:19:37.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nI addressed the comments in https://review.opendev.org/741925","accounts_in_message":[],"_revision_number":5},{"id":"61570b23367bea4e7bf7416841def0e2b3f9503f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-20 13:34:00.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(2 comments)\n\nlgtm, let\u0027s do the improvement in followup.","accounts_in_message":[],"_revision_number":5},{"id":"15b78ce0cb0f5f51a98b98ad1bd145617f95944a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 13:34:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"76041c5b402a12cc1e54368fb39db12bc1665765","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 13:39:00.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac7dc5d6cb548509e9fecf2666465f4 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":5},{"id":"1099ce226cb4a6074c54de6c88226f68d3be0368","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 13:39:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"c96a528a972933000c7257da2cd33059ae55d1bd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 13:41:26.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/8815a4cbe7cb4a3c938c9de159b594fe : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"5a8b34fbba7c0744456f5d32167e0295f8578387","revisions":{"836c9de74760d4817f909ac1d9e8391ad9712292":{"kind":"REWORK","_number":1,"created":"2020-07-13 16:28:53.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/27/740727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/27/740727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/1"}}},"commit":{"parents":[{"commit":"647bcbdc1cc6456a09a9580b58e51a32fbe7f567","subject":"Update release management page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/647bcbdc1cc6456a09a9580b58e51a32fbe7f567"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-13 16:04:01.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-13 16:28:50.000000000","tz":120},"subject":"Import the Zuul v3 porting guide","message":"Import the Zuul v3 porting guide\n\nThe guide needs a reference location. It was removed from\nopendev/infra-manual via https://review.opendev.org/711325.\nThis change adds a bit of introduction and and imports\nthe last version of the guide almost unchanged:\n- minimal changes to few then-internal references, now external;\n- a few formatting fixes (long lines).\n\nThis is a squashed version of zuulv3.rst which comes from\nthe following Change-Id-s:\n\nI0db0a753e17d671ee64ea7a87e065c958dd94850\nI2a5fb17c5ec1cb76693bf55ff25ffda6cc84759e\nI4bfb63bf99c8528494d73f5f50018a2636b5bc37\nIe79726677c3c3ad5204fa6c1b3d3c0dfa25d7ebb\nI720e1532f177d26fbb50f00dbc612422aa60c3bd\nI51bd6c5bbcb8135c3ea4fed6941613643a3bcead\nI0f8878901ce0302c89a40b9190dd806b2f5a85e8\nIc08e914be065ff9f4ec6b489c9d1d9915ff284f8\nI837ba17307cd81b4339c2a1350692e3186146fb3\nI33bab470f04a00b978b8cf22416ac4d0d805b4ef\nIdded1004ffaebe51010fa52dad696f19f721d526\nIbe8dfe1ce3723dee8d682a7d3fa4cb57561614d9\nIea9b8386e22c8801343e5405d7a1a249a93df7d9\nId02e8c89fd212d527f8efcc575eb5541ed06a3ad\nI2b4b4a9c3d9009cc618b7ebba666582e4cf03a08\nI2b150ccf7e712ddfaf9ab61a20e986630ce94501\nIc372436b97fc26173767033346172883c2f91a37\nI5e69c55b9b7ea2c3654f4a231005107f7ead0430\nI084c0e930fb4b7899e4d233f9529642bbcf19a7d\nIf36eed0a720f3e4747b81359687a9e5df417c0fe\nIbf79b03c33fb82d2e55016e1e1225b0d9dba6f1e\nIb3102f270f94eb689b81dd719574b1113bccba19\nIb78c43e4a38c3976ca9180fb68bbfb5eedaf3df7\nIf30ddf3f32bc49e8c764b123ac760a34bcae9cec\nI2e099e0d90e0450fba89ac985de3aa646836367d\nI0d3a3dfe4015dc8ea9f7d3957fe5de29c3e2283b\nI3156520772db4de7cc90fcfc274b3dac957538e4\nI09a0f9ecb3bba693caac67958675cce7fd554a5e\nIa2b71fbd2cf5a7848db1648278350c91e757a324\nIb8062210854e1979aa1f56bd7c0f5af8e578decc\nI79ec68b995c5acc431b6cc8007ffd00808c89ff0\nI112005d360f23648985d6691200d07476c622d70\nI1649c64e2e7b2bd9b6324f44a8a50bce6f2fdb98\nIe85c7ebc1a1c9f646316e8c7e6d6eb7eea8a23d3\nI73970d0b70c1de919e28d669a531e6d6d2bbef71\nI8bbf3d937b71847ea499539b50283bb1f255d1f3\nI21dbdb66962b44eba460300ade8f28869f8b9c07\nIacd4f3057c3e33396f6930283177460624ff102c\nI52417a26587b42c4e18a5d81dbb84643f1b36115\nIe2f37c17747e1f22e517457e39e3721d00d27ab6\nIa093e69ca515895221258240d74d30065c76eabf\nIc690f2e680992d365c760ca8f94b85d7ba78bbc7\nIe313d59dfaf1a0ab1a5a87f53e270d40c6206c21\nI68e9320570227ce53bae422cd93253f8d561cb66\nI542c8f7cd383415b0b1e421ad68685d62ebed9a1\nI2f720a9e6ff3c10ac05852d7d102878f418bd5cf\nId44c3eda204456af3dfa6d616ff44c107088fb40\nIe2f414d271bc1ee649b50e24cc3aa6bc80c973c6\nI19493c9fb9f07561c9627e080ed24c0a58164a53\nI7b467cb8b13e614d382c4801aa244e34ce827065\nIeb1f88e07eb0660f62699819d4be833d5d794e65\nIdb46fc3c6fc07aad8fa2f4e20b43785bccd90547\nIafad5ae4997c94e1c517bab1fb794ff5d6a34251\nIa4ee7f2c8c623ea0054c41007b16aeed65d13e35\n\nChange-Id: Ib33823e449ab98455f83a5be703b69ca3b886039\nCo-Authored-By: James E. Blair \u003cjeblair@redhat.com\u003e\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\nCo-Authored-By: Monty Taylor \u003cmordred@inaugust.com\u003e\nCo-Authored-By: David Shrewsbury \u003cshrewsbury.dave@gmail.com\u003e\nCo-Authored-By: Clark Boylan \u003cclark.boylan@gmail.com\u003e\nCo-Authored-By: David Moreau-Simard \u003cdmsimard@redhat.com\u003e\nCo-Authored-By: Doug Hellmann \u003cdoug@doughellmann.com\u003e\nCo-Authored-By: François Magimel \u003cfrancois.magimel@alumni.enseeiht.fr\u003e\nCo-Authored-By: Jan Zerebecki \u003cjan.openstack@zerebecki.de\u003e\nCo-Authored-By: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nCo-Authored-By: Jesse Keating \u003comgjlk@us.ibm.com\u003e\nCo-Authored-By: Julien Danjou \u003cjulien@danjou.info\u003e\nCo-Authored-By: Nguyen Van Trung \u003ctrungnv@vn.fujitsu.com\u003e\nCo-Authored-By: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCo-Authored-By: YAMAMOTO Takashi \u003cyamamoto@midokura.com\u003e\nCo-Authored-By: melissaml \u003cma.lei@99cloud.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/836c9de74760d4817f909ac1d9e8391ad9712292"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/836c9de74760d4817f909ac1d9e8391ad9712292"}]},"branch":"refs/heads/master"},"ac6a37aeb60433a9b29188e649fece3cab845aba":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-13 16:29:49.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/27/740727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/27/740727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/2"}}},"commit":{"parents":[{"commit":"647bcbdc1cc6456a09a9580b58e51a32fbe7f567","subject":"Update release management page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/647bcbdc1cc6456a09a9580b58e51a32fbe7f567"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-13 16:29:36.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-13 16:29:36.000000000","tz":120},"subject":"Import the Zuul v3 porting guide","message":"Import the Zuul v3 porting guide\n\nThe guide needs a reference location. It was removed from\nopendev/infra-manual via https://review.opendev.org/711325.\nThis change adds a bit of introduction and and imports\nthe last version of the guide almost unchanged:\n- minimal changes to few then-internal references, now external;\n- a few formatting fixes (long lines).\n\nThis is a squashed version of zuulv3.rst which comes from\nthe following Change-Id-s:\n\nI0db0a753e17d671ee64ea7a87e065c958dd94850\nI2a5fb17c5ec1cb76693bf55ff25ffda6cc84759e\nI4bfb63bf99c8528494d73f5f50018a2636b5bc37\nIe79726677c3c3ad5204fa6c1b3d3c0dfa25d7ebb\nI720e1532f177d26fbb50f00dbc612422aa60c3bd\nI51bd6c5bbcb8135c3ea4fed6941613643a3bcead\nI0f8878901ce0302c89a40b9190dd806b2f5a85e8\nIc08e914be065ff9f4ec6b489c9d1d9915ff284f8\nI837ba17307cd81b4339c2a1350692e3186146fb3\nI33bab470f04a00b978b8cf22416ac4d0d805b4ef\nIdded1004ffaebe51010fa52dad696f19f721d526\nIbe8dfe1ce3723dee8d682a7d3fa4cb57561614d9\nIea9b8386e22c8801343e5405d7a1a249a93df7d9\nId02e8c89fd212d527f8efcc575eb5541ed06a3ad\nI2b4b4a9c3d9009cc618b7ebba666582e4cf03a08\nI2b150ccf7e712ddfaf9ab61a20e986630ce94501\nIc372436b97fc26173767033346172883c2f91a37\nI5e69c55b9b7ea2c3654f4a231005107f7ead0430\nI084c0e930fb4b7899e4d233f9529642bbcf19a7d\nIf36eed0a720f3e4747b81359687a9e5df417c0fe\nIbf79b03c33fb82d2e55016e1e1225b0d9dba6f1e\nIb3102f270f94eb689b81dd719574b1113bccba19\nIb78c43e4a38c3976ca9180fb68bbfb5eedaf3df7\nIf30ddf3f32bc49e8c764b123ac760a34bcae9cec\nI2e099e0d90e0450fba89ac985de3aa646836367d\nI0d3a3dfe4015dc8ea9f7d3957fe5de29c3e2283b\nI3156520772db4de7cc90fcfc274b3dac957538e4\nI09a0f9ecb3bba693caac67958675cce7fd554a5e\nIa2b71fbd2cf5a7848db1648278350c91e757a324\nIb8062210854e1979aa1f56bd7c0f5af8e578decc\nI79ec68b995c5acc431b6cc8007ffd00808c89ff0\nI112005d360f23648985d6691200d07476c622d70\nI1649c64e2e7b2bd9b6324f44a8a50bce6f2fdb98\nIe85c7ebc1a1c9f646316e8c7e6d6eb7eea8a23d3\nI73970d0b70c1de919e28d669a531e6d6d2bbef71\nI8bbf3d937b71847ea499539b50283bb1f255d1f3\nI21dbdb66962b44eba460300ade8f28869f8b9c07\nIacd4f3057c3e33396f6930283177460624ff102c\nI52417a26587b42c4e18a5d81dbb84643f1b36115\nIe2f37c17747e1f22e517457e39e3721d00d27ab6\nIa093e69ca515895221258240d74d30065c76eabf\nIc690f2e680992d365c760ca8f94b85d7ba78bbc7\nIe313d59dfaf1a0ab1a5a87f53e270d40c6206c21\nI68e9320570227ce53bae422cd93253f8d561cb66\nI542c8f7cd383415b0b1e421ad68685d62ebed9a1\nI2f720a9e6ff3c10ac05852d7d102878f418bd5cf\nId44c3eda204456af3dfa6d616ff44c107088fb40\nIe2f414d271bc1ee649b50e24cc3aa6bc80c973c6\nI19493c9fb9f07561c9627e080ed24c0a58164a53\nI7b467cb8b13e614d382c4801aa244e34ce827065\nIeb1f88e07eb0660f62699819d4be833d5d794e65\nIdb46fc3c6fc07aad8fa2f4e20b43785bccd90547\nIafad5ae4997c94e1c517bab1fb794ff5d6a34251\nIa4ee7f2c8c623ea0054c41007b16aeed65d13e35\n\nCo-Authored-By: James E. Blair \u003cjeblair@redhat.com\u003e\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\nCo-Authored-By: Monty Taylor \u003cmordred@inaugust.com\u003e\nCo-Authored-By: David Shrewsbury \u003cshrewsbury.dave@gmail.com\u003e\nCo-Authored-By: Clark Boylan \u003cclark.boylan@gmail.com\u003e\nCo-Authored-By: David Moreau-Simard \u003cdmsimard@redhat.com\u003e\nCo-Authored-By: Doug Hellmann \u003cdoug@doughellmann.com\u003e\nCo-Authored-By: François Magimel \u003cfrancois.magimel@alumni.enseeiht.fr\u003e\nCo-Authored-By: Jan Zerebecki \u003cjan.openstack@zerebecki.de\u003e\nCo-Authored-By: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nCo-Authored-By: Jesse Keating \u003comgjlk@us.ibm.com\u003e\nCo-Authored-By: Julien Danjou \u003cjulien@danjou.info\u003e\nCo-Authored-By: Nguyen Van Trung \u003ctrungnv@vn.fujitsu.com\u003e\nCo-Authored-By: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCo-Authored-By: YAMAMOTO Takashi \u003cyamamoto@midokura.com\u003e\nCo-Authored-By: melissaml \u003cma.lei@99cloud.net\u003e\n\nChange-Id: Ib33823e449ab98455f83a5be703b69ca3b886039\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ac6a37aeb60433a9b29188e649fece3cab845aba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ac6a37aeb60433a9b29188e649fece3cab845aba"}]},"branch":"refs/heads/master"},"96a91d5c8de32046da1dce868afe786748eab9dc":{"kind":"REWORK","_number":3,"created":"2020-07-13 17:25:54.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/27/740727/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/27/740727/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/3"}}},"commit":{"parents":[{"commit":"647bcbdc1cc6456a09a9580b58e51a32fbe7f567","subject":"Update release management page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/647bcbdc1cc6456a09a9580b58e51a32fbe7f567"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-13 17:25:45.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-13 17:25:45.000000000","tz":120},"subject":"Import the Zuul v3 porting guide","message":"Import the Zuul v3 porting guide\n\nThe guide needs a reference location. It was removed from\nopendev/infra-manual via https://review.opendev.org/711325.\nThis change adds a bit of introduction and and imports\nthe last version of the guide almost unchanged:\n- minimal changes to few then-internal references, now external;\n- a few formatting fixes (long lines).\n\nThis is a squashed version of zuulv3.rst which comes from\nthe following Change-Id-s:\n\nI0db0a753e17d671ee64ea7a87e065c958dd94850\nI2a5fb17c5ec1cb76693bf55ff25ffda6cc84759e\nI4bfb63bf99c8528494d73f5f50018a2636b5bc37\nIe79726677c3c3ad5204fa6c1b3d3c0dfa25d7ebb\nI720e1532f177d26fbb50f00dbc612422aa60c3bd\nI51bd6c5bbcb8135c3ea4fed6941613643a3bcead\nI0f8878901ce0302c89a40b9190dd806b2f5a85e8\nIc08e914be065ff9f4ec6b489c9d1d9915ff284f8\nI837ba17307cd81b4339c2a1350692e3186146fb3\nI33bab470f04a00b978b8cf22416ac4d0d805b4ef\nIdded1004ffaebe51010fa52dad696f19f721d526\nIbe8dfe1ce3723dee8d682a7d3fa4cb57561614d9\nIea9b8386e22c8801343e5405d7a1a249a93df7d9\nId02e8c89fd212d527f8efcc575eb5541ed06a3ad\nI2b4b4a9c3d9009cc618b7ebba666582e4cf03a08\nI2b150ccf7e712ddfaf9ab61a20e986630ce94501\nIc372436b97fc26173767033346172883c2f91a37\nI5e69c55b9b7ea2c3654f4a231005107f7ead0430\nI084c0e930fb4b7899e4d233f9529642bbcf19a7d\nIf36eed0a720f3e4747b81359687a9e5df417c0fe\nIbf79b03c33fb82d2e55016e1e1225b0d9dba6f1e\nIb3102f270f94eb689b81dd719574b1113bccba19\nIb78c43e4a38c3976ca9180fb68bbfb5eedaf3df7\nIf30ddf3f32bc49e8c764b123ac760a34bcae9cec\nI2e099e0d90e0450fba89ac985de3aa646836367d\nI0d3a3dfe4015dc8ea9f7d3957fe5de29c3e2283b\nI3156520772db4de7cc90fcfc274b3dac957538e4\nI09a0f9ecb3bba693caac67958675cce7fd554a5e\nIa2b71fbd2cf5a7848db1648278350c91e757a324\nIb8062210854e1979aa1f56bd7c0f5af8e578decc\nI79ec68b995c5acc431b6cc8007ffd00808c89ff0\nI112005d360f23648985d6691200d07476c622d70\nI1649c64e2e7b2bd9b6324f44a8a50bce6f2fdb98\nIe85c7ebc1a1c9f646316e8c7e6d6eb7eea8a23d3\nI73970d0b70c1de919e28d669a531e6d6d2bbef71\nI8bbf3d937b71847ea499539b50283bb1f255d1f3\nI21dbdb66962b44eba460300ade8f28869f8b9c07\nIacd4f3057c3e33396f6930283177460624ff102c\nI52417a26587b42c4e18a5d81dbb84643f1b36115\nIe2f37c17747e1f22e517457e39e3721d00d27ab6\nIa093e69ca515895221258240d74d30065c76eabf\nIc690f2e680992d365c760ca8f94b85d7ba78bbc7\nIe313d59dfaf1a0ab1a5a87f53e270d40c6206c21\nI68e9320570227ce53bae422cd93253f8d561cb66\nI542c8f7cd383415b0b1e421ad68685d62ebed9a1\nI2f720a9e6ff3c10ac05852d7d102878f418bd5cf\nId44c3eda204456af3dfa6d616ff44c107088fb40\nIe2f414d271bc1ee649b50e24cc3aa6bc80c973c6\nI19493c9fb9f07561c9627e080ed24c0a58164a53\nI7b467cb8b13e614d382c4801aa244e34ce827065\nIeb1f88e07eb0660f62699819d4be833d5d794e65\nIdb46fc3c6fc07aad8fa2f4e20b43785bccd90547\nIafad5ae4997c94e1c517bab1fb794ff5d6a34251\nIa4ee7f2c8c623ea0054c41007b16aeed65d13e35\n\nCo-Authored-By: James E. Blair \u003cjeblair@redhat.com\u003e\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\nCo-Authored-By: Monty Taylor \u003cmordred@inaugust.com\u003e\nCo-Authored-By: David Shrewsbury \u003cshrewsbury.dave@gmail.com\u003e\nCo-Authored-By: Clark Boylan \u003cclark.boylan@gmail.com\u003e\nCo-Authored-By: David Moreau-Simard \u003cdmsimard@redhat.com\u003e\nCo-Authored-By: Doug Hellmann \u003cdoug@doughellmann.com\u003e\nCo-Authored-By: François Magimel \u003cfrancois.magimel@alumni.enseeiht.fr\u003e\nCo-Authored-By: Jan Zerebecki \u003cjan.openstack@zerebecki.de\u003e\nCo-Authored-By: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nCo-Authored-By: Jesse Keating \u003comgjlk@us.ibm.com\u003e\nCo-Authored-By: Julien Danjou \u003cjulien@danjou.info\u003e\nCo-Authored-By: Nguyen Van Trung \u003ctrungnv@vn.fujitsu.com\u003e\nCo-Authored-By: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCo-Authored-By: YAMAMOTO Takashi \u003cyamamoto@midokura.com\u003e\nCo-Authored-By: melissaml \u003cma.lei@99cloud.net\u003e\n\nChange-Id: Ib33823e449ab98455f83a5be703b69ca3b886039\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/96a91d5c8de32046da1dce868afe786748eab9dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/96a91d5c8de32046da1dce868afe786748eab9dc"}]},"branch":"refs/heads/master"},"48226164ae02ac3c98fb6416592ff8fed3c27e4b":{"kind":"REWORK","_number":4,"created":"2020-07-14 08:46:38.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/27/740727/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/27/740727/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/4"}}},"commit":{"parents":[{"commit":"647bcbdc1cc6456a09a9580b58e51a32fbe7f567","subject":"Update release management page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/647bcbdc1cc6456a09a9580b58e51a32fbe7f567"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-14 08:46:15.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-14 08:46:15.000000000","tz":120},"subject":"Import the Zuul v3 porting guide","message":"Import the Zuul v3 porting guide\n\nThe guide needs a reference location. It was removed from\nopendev/infra-manual via https://review.opendev.org/711325.\nThis change adds a bit of introduction and and imports\nthe last version of the guide almost unchanged:\n- minimal changes to few then-internal references, now external;\n- a few formatting fixes (long lines, broken links, etc).\n\nThis is a squashed version of zuulv3.rst which comes from\nthe following Change-Id-s:\n\nI0db0a753e17d671ee64ea7a87e065c958dd94850\nI2a5fb17c5ec1cb76693bf55ff25ffda6cc84759e\nI4bfb63bf99c8528494d73f5f50018a2636b5bc37\nIe79726677c3c3ad5204fa6c1b3d3c0dfa25d7ebb\nI720e1532f177d26fbb50f00dbc612422aa60c3bd\nI51bd6c5bbcb8135c3ea4fed6941613643a3bcead\nI0f8878901ce0302c89a40b9190dd806b2f5a85e8\nIc08e914be065ff9f4ec6b489c9d1d9915ff284f8\nI837ba17307cd81b4339c2a1350692e3186146fb3\nI33bab470f04a00b978b8cf22416ac4d0d805b4ef\nIdded1004ffaebe51010fa52dad696f19f721d526\nIbe8dfe1ce3723dee8d682a7d3fa4cb57561614d9\nIea9b8386e22c8801343e5405d7a1a249a93df7d9\nId02e8c89fd212d527f8efcc575eb5541ed06a3ad\nI2b4b4a9c3d9009cc618b7ebba666582e4cf03a08\nI2b150ccf7e712ddfaf9ab61a20e986630ce94501\nIc372436b97fc26173767033346172883c2f91a37\nI5e69c55b9b7ea2c3654f4a231005107f7ead0430\nI084c0e930fb4b7899e4d233f9529642bbcf19a7d\nIf36eed0a720f3e4747b81359687a9e5df417c0fe\nIbf79b03c33fb82d2e55016e1e1225b0d9dba6f1e\nIb3102f270f94eb689b81dd719574b1113bccba19\nIb78c43e4a38c3976ca9180fb68bbfb5eedaf3df7\nIf30ddf3f32bc49e8c764b123ac760a34bcae9cec\nI2e099e0d90e0450fba89ac985de3aa646836367d\nI0d3a3dfe4015dc8ea9f7d3957fe5de29c3e2283b\nI3156520772db4de7cc90fcfc274b3dac957538e4\nI09a0f9ecb3bba693caac67958675cce7fd554a5e\nIa2b71fbd2cf5a7848db1648278350c91e757a324\nIb8062210854e1979aa1f56bd7c0f5af8e578decc\nI79ec68b995c5acc431b6cc8007ffd00808c89ff0\nI112005d360f23648985d6691200d07476c622d70\nI1649c64e2e7b2bd9b6324f44a8a50bce6f2fdb98\nIe85c7ebc1a1c9f646316e8c7e6d6eb7eea8a23d3\nI73970d0b70c1de919e28d669a531e6d6d2bbef71\nI8bbf3d937b71847ea499539b50283bb1f255d1f3\nI21dbdb66962b44eba460300ade8f28869f8b9c07\nIacd4f3057c3e33396f6930283177460624ff102c\nI52417a26587b42c4e18a5d81dbb84643f1b36115\nIe2f37c17747e1f22e517457e39e3721d00d27ab6\nIa093e69ca515895221258240d74d30065c76eabf\nIc690f2e680992d365c760ca8f94b85d7ba78bbc7\nIe313d59dfaf1a0ab1a5a87f53e270d40c6206c21\nI68e9320570227ce53bae422cd93253f8d561cb66\nI542c8f7cd383415b0b1e421ad68685d62ebed9a1\nI2f720a9e6ff3c10ac05852d7d102878f418bd5cf\nId44c3eda204456af3dfa6d616ff44c107088fb40\nIe2f414d271bc1ee649b50e24cc3aa6bc80c973c6\nI19493c9fb9f07561c9627e080ed24c0a58164a53\nI7b467cb8b13e614d382c4801aa244e34ce827065\nIeb1f88e07eb0660f62699819d4be833d5d794e65\nIdb46fc3c6fc07aad8fa2f4e20b43785bccd90547\nIafad5ae4997c94e1c517bab1fb794ff5d6a34251\nIa4ee7f2c8c623ea0054c41007b16aeed65d13e35\n\nCo-Authored-By: James E. Blair \u003cjeblair@redhat.com\u003e\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\nCo-Authored-By: Monty Taylor \u003cmordred@inaugust.com\u003e\nCo-Authored-By: David Shrewsbury \u003cshrewsbury.dave@gmail.com\u003e\nCo-Authored-By: Clark Boylan \u003cclark.boylan@gmail.com\u003e\nCo-Authored-By: David Moreau-Simard \u003cdmsimard@redhat.com\u003e\nCo-Authored-By: Doug Hellmann \u003cdoug@doughellmann.com\u003e\nCo-Authored-By: François Magimel \u003cfrancois.magimel@alumni.enseeiht.fr\u003e\nCo-Authored-By: Jan Zerebecki \u003cjan.openstack@zerebecki.de\u003e\nCo-Authored-By: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nCo-Authored-By: Jesse Keating \u003comgjlk@us.ibm.com\u003e\nCo-Authored-By: Julien Danjou \u003cjulien@danjou.info\u003e\nCo-Authored-By: Nguyen Van Trung \u003ctrungnv@vn.fujitsu.com\u003e\nCo-Authored-By: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCo-Authored-By: YAMAMOTO Takashi \u003cyamamoto@midokura.com\u003e\nCo-Authored-By: melissaml \u003cma.lei@99cloud.net\u003e\n\nChange-Id: Ib33823e449ab98455f83a5be703b69ca3b886039\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/48226164ae02ac3c98fb6416592ff8fed3c27e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/48226164ae02ac3c98fb6416592ff8fed3c27e4b"}]},"branch":"refs/heads/master"},"5a8b34fbba7c0744456f5d32167e0295f8578387":{"kind":"REWORK","_number":5,"created":"2020-07-14 11:26:35.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/27/740727/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/27/740727/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/27/740727/5"}}},"commit":{"parents":[{"commit":"647bcbdc1cc6456a09a9580b58e51a32fbe7f567","subject":"Update release management page","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/647bcbdc1cc6456a09a9580b58e51a32fbe7f567"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-14 11:26:27.000000000","tz":120},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2020-07-14 11:26:27.000000000","tz":120},"subject":"Import the Zuul v3 porting guide","message":"Import the Zuul v3 porting guide\n\nThe guide needs a reference location. It was removed from\nopendev/infra-manual via https://review.opendev.org/711325.\nThis change adds a bit of introduction and and imports\nthe last version of the guide almost unchanged:\n- minimal changes to few then-internal references, now external;\n- a few formatting fixes (long lines, broken links, etc).\n\nThis is a squashed version of zuulv3.rst which comes from\nthe following Change-Id-s:\n\nI0db0a753e17d671ee64ea7a87e065c958dd94850\nI2a5fb17c5ec1cb76693bf55ff25ffda6cc84759e\nI4bfb63bf99c8528494d73f5f50018a2636b5bc37\nIe79726677c3c3ad5204fa6c1b3d3c0dfa25d7ebb\nI720e1532f177d26fbb50f00dbc612422aa60c3bd\nI51bd6c5bbcb8135c3ea4fed6941613643a3bcead\nI0f8878901ce0302c89a40b9190dd806b2f5a85e8\nIc08e914be065ff9f4ec6b489c9d1d9915ff284f8\nI837ba17307cd81b4339c2a1350692e3186146fb3\nI33bab470f04a00b978b8cf22416ac4d0d805b4ef\nIdded1004ffaebe51010fa52dad696f19f721d526\nIbe8dfe1ce3723dee8d682a7d3fa4cb57561614d9\nIea9b8386e22c8801343e5405d7a1a249a93df7d9\nId02e8c89fd212d527f8efcc575eb5541ed06a3ad\nI2b4b4a9c3d9009cc618b7ebba666582e4cf03a08\nI2b150ccf7e712ddfaf9ab61a20e986630ce94501\nIc372436b97fc26173767033346172883c2f91a37\nI5e69c55b9b7ea2c3654f4a231005107f7ead0430\nI084c0e930fb4b7899e4d233f9529642bbcf19a7d\nIf36eed0a720f3e4747b81359687a9e5df417c0fe\nIbf79b03c33fb82d2e55016e1e1225b0d9dba6f1e\nIb3102f270f94eb689b81dd719574b1113bccba19\nIb78c43e4a38c3976ca9180fb68bbfb5eedaf3df7\nIf30ddf3f32bc49e8c764b123ac760a34bcae9cec\nI2e099e0d90e0450fba89ac985de3aa646836367d\nI0d3a3dfe4015dc8ea9f7d3957fe5de29c3e2283b\nI3156520772db4de7cc90fcfc274b3dac957538e4\nI09a0f9ecb3bba693caac67958675cce7fd554a5e\nIa2b71fbd2cf5a7848db1648278350c91e757a324\nIb8062210854e1979aa1f56bd7c0f5af8e578decc\nI79ec68b995c5acc431b6cc8007ffd00808c89ff0\nI112005d360f23648985d6691200d07476c622d70\nI1649c64e2e7b2bd9b6324f44a8a50bce6f2fdb98\nIe85c7ebc1a1c9f646316e8c7e6d6eb7eea8a23d3\nI73970d0b70c1de919e28d669a531e6d6d2bbef71\nI8bbf3d937b71847ea499539b50283bb1f255d1f3\nI21dbdb66962b44eba460300ade8f28869f8b9c07\nIacd4f3057c3e33396f6930283177460624ff102c\nI52417a26587b42c4e18a5d81dbb84643f1b36115\nIe2f37c17747e1f22e517457e39e3721d00d27ab6\nIa093e69ca515895221258240d74d30065c76eabf\nIc690f2e680992d365c760ca8f94b85d7ba78bbc7\nIe313d59dfaf1a0ab1a5a87f53e270d40c6206c21\nI68e9320570227ce53bae422cd93253f8d561cb66\nI542c8f7cd383415b0b1e421ad68685d62ebed9a1\nI2f720a9e6ff3c10ac05852d7d102878f418bd5cf\nId44c3eda204456af3dfa6d616ff44c107088fb40\nIe2f414d271bc1ee649b50e24cc3aa6bc80c973c6\nI19493c9fb9f07561c9627e080ed24c0a58164a53\nI7b467cb8b13e614d382c4801aa244e34ce827065\nIeb1f88e07eb0660f62699819d4be833d5d794e65\nIdb46fc3c6fc07aad8fa2f4e20b43785bccd90547\nIafad5ae4997c94e1c517bab1fb794ff5d6a34251\nIa4ee7f2c8c623ea0054c41007b16aeed65d13e35\n\nCo-Authored-By: James E. Blair \u003cjeblair@redhat.com\u003e\nCo-Authored-By: Andreas Jaeger \u003caj@suse.com\u003e\nCo-Authored-By: Monty Taylor \u003cmordred@inaugust.com\u003e\nCo-Authored-By: David Shrewsbury \u003cshrewsbury.dave@gmail.com\u003e\nCo-Authored-By: Clark Boylan \u003cclark.boylan@gmail.com\u003e\nCo-Authored-By: David Moreau-Simard \u003cdmsimard@redhat.com\u003e\nCo-Authored-By: Doug Hellmann \u003cdoug@doughellmann.com\u003e\nCo-Authored-By: François Magimel \u003cfrancois.magimel@alumni.enseeiht.fr\u003e\nCo-Authored-By: Jan Zerebecki \u003cjan.openstack@zerebecki.de\u003e\nCo-Authored-By: Jeremy Stanley \u003cfungi@yuggoth.org\u003e\nCo-Authored-By: Jesse Keating \u003comgjlk@us.ibm.com\u003e\nCo-Authored-By: Julien Danjou \u003cjulien@danjou.info\u003e\nCo-Authored-By: Nguyen Van Trung \u003ctrungnv@vn.fujitsu.com\u003e\nCo-Authored-By: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCo-Authored-By: YAMAMOTO Takashi \u003cyamamoto@midokura.com\u003e\nCo-Authored-By: melissaml \u003cma.lei@99cloud.net\u003e\n\nChange-Id: Ib33823e449ab98455f83a5be703b69ca3b886039\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/5a8b34fbba7c0744456f5d32167e0295f8578387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/5a8b34fbba7c0744456f5d32167e0295f8578387"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
