)]}'
{"id":"openstack%2Fproject-team-guide~810499","triplet_id":"openstack%2Fproject-team-guide~master~I9589cbdd74d83d23ad42ce6724e3b8e4cc56c44d","project":"openstack/project-team-guide","branch":"master","hashtags":[],"change_id":"I9589cbdd74d83d23ad42ce6724e3b8e4cc56c44d","subject":"Remove \u0027Unmaintained\u0027 phase","status":"MERGED","created":"2021-09-22 18:13:29.000000000","updated":"2021-09-28 22:00:59.000000000","submitted":"2021-09-28 21:59:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"810499","meta_rev_id":"97ece86aba366f261c802dfd7d452fe18d393c67","_number":810499,"virtual_id_number":810499,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-28 21:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-09-23 16:33:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-09-23 08:27:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2021-09-25 15:46:04.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2021-09-23 13:56:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-22 19:08:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-09-27 15:28:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-22 18:15:56.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-09-22 18:21:33.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2021-09-22 19:08:48.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-09-23 08:27:50.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-09-23 16:33:11.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-09-25 15:46:04.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-09-28 21:51:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bda37a9e0d9fd2c39eaf5fa5612a051a4a2cb8ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-22 18:13:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5abad5c34aaf4784ec2b8cc87c5e31d682ed252d","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-09-22 18:21:33.000000000","message":"Patch Set 1:\n\nI\u0027m tentatively okay with this, but remember that discussions which happen at the PTG aren\u0027t necessarily inclusive of the larger community, so linking to where it was discussed on the mailing list (it *was* discussed on the ML, right?) would be a good addition to the commit message for posterity.\n\nMore generally, the \"unmaintained\" phase was intended to encourage teams to have a waiting period between when there was nobody to take care of backports or fix CI jobs and when the EOL tag got applied and the branch deleted. This was meant as a warning to anyone consuming that stable branch that if they didn\u0027t pitch in and help maintain it, then it would be gone permanently, since we don\u0027t re-create branches after EOL. I think getting rid of that warning period as an explicit phase is fine, but the guide could stand to recommend to projects considering EOL of an EM branch that they first provide ample warning to the broader community of the situation, in case there\u0027s someone interested in helping reverse course away from EOL before it\u0027s too late.","accounts_in_message":[],"_revision_number":1},{"id":"3c3041020470fcb79f1cc3da70b18a316883c8ad","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-22 18:39:34.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e Patch Set 1:\n\u003e \n\u003e I\u0027m tentatively okay with this, but remember that discussions which happen at the PTG aren\u0027t necessarily inclusive of the larger community, so linking to where it was discussed on the mailing list (it *was* discussed on the ML, right?) would be a good addition to the commit message for posterity.\n\u003e \n\u003e More generally, the \"unmaintained\" phase was intended to encourage teams to have a waiting period between when there was nobody to take care of backports or fix CI jobs and when the EOL tag got applied and the branch deleted. This was meant as a warning to anyone consuming that stable branch that if they didn\u0027t pitch in and help maintain it, then it would be gone permanently, since we don\u0027t re-create branches after EOL. I think getting rid of that warning period as an explicit phase is fine, but the guide could stand to recommend to projects considering EOL of an EM branch that they first provide ample warning to the broader community of the situation, in case there\u0027s someone interested in helping reverse course away from EOL before it\u0027s too late.\n\n\nThis \u0027unmaintained\u0027 phase is not really used for anything since it is created. we saw stable directly going from EM to EOL like stable/ocata. I feel we do not need this as such and for warning there is a mandatory step for PTL/projects to send warning to ML before they do transition:\n\n\"To initiate this transition, either the PTL of the given project or other\nstable maintainer should:\n\n#. Send an announcement to the openstack-discuss mailing list (in order to give\n   some time for others to step up as maintainers if there are volunteers).\n.....\"\n\n\nI am ok with this but let\u0027s send it on ML also about what we agreed  TC PTG session.","accounts_in_message":[],"_revision_number":1},{"id":"f78b8d396d46947cb46fddd22e51607df290f063","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-22 18:44:30.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e Patch Set 1:\n\u003e \n\u003e I\u0027m tentatively okay with this, but remember that discussions which happen at the PTG aren\u0027t necessarily inclusive of the larger community, so linking to where it was discussed on the mailing list (it *was* discussed on the ML, right?) would be a good addition to the commit message for posterity.\n\u003e \n\u003e More generally, the \"unmaintained\" phase was intended to encourage teams to have a waiting period between when there was nobody to take care of backports or fix CI jobs and when the EOL tag got applied and the branch deleted. This was meant as a warning to anyone consuming that stable branch that if they didn\u0027t pitch in and help maintain it, then it would be gone permanently, since we don\u0027t re-create branches after EOL. I think getting rid of that warning period as an explicit phase is fine, but the guide could stand to recommend to projects considering EOL of an EM branch that they first provide ample warning to the broader community of the situation, in case there\u0027s someone interested in helping reverse course away from EOL before it\u0027s too late.\n\nThank you for the review and thanks for the correct explanation of the original purpose of the unmaintained phase fungi! :)\n\nWell, we most probably touched this in the ML as well in the past, I\u0027ll try to look for this topic in mails. On the other hand, yes, this was discussed on the PTG, and as a result an AGREEMENT was added [1].\n\n@fungi: Do you think that we would need to advertise this change on ML now as well?\n\n[1] Line 397: https://etherpad.opendev.org/p/tc-xena-ptg","accounts_in_message":[],"_revision_number":1},{"id":"10ed75c95d811fc0b828fc6cf77107d600c53912","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-22 18:45:51.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+1\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e I\u0027m tentatively okay with this, but remember that discussions which happen at the PTG aren\u0027t necessarily inclusive of the larger community, so linking to where it was discussed on the mailing list (it *was* discussed on the ML, right?) would be a good addition to the commit message for posterity.\n\u003e \u003e \n\u003e \u003e More generally, the \"unmaintained\" phase was intended to encourage teams to have a waiting period between when there was nobody to take care of backports or fix CI jobs and when the EOL tag got applied and the branch deleted. This was meant as a warning to anyone consuming that stable branch that if they didn\u0027t pitch in and help maintain it, then it would be gone permanently, since we don\u0027t re-create branches after EOL. I think getting rid of that warning period as an explicit phase is fine, but the guide could stand to recommend to projects considering EOL of an EM branch that they first provide ample warning to the broader community of the situation, in case there\u0027s someone interested in helping reverse course away from EOL before it\u0027s too late.\n\u003e \n\u003e \n\u003e This \u0027unmaintained\u0027 phase is not really used for anything since it is created. we saw stable directly going from EM to EOL like stable/ocata. I feel we do not need this as such and for warning there is a mandatory step for PTL/projects to send warning to ML before they do transition:\n\u003e \n\u003e \"To initiate this transition, either the PTL of the given project or other\n\u003e stable maintainer should:\n\u003e \n\u003e #. Send an announcement to the openstack-discuss mailing list (in order to give\n\u003e    some time for others to step up as maintainers if there are volunteers).\n\u003e .....\"\n\u003e \n\u003e \n\u003e I am ok with this but let\u0027s send it on ML also about what we agreed  TC PTG session.\n\nOk, thank you Ghanshyam! I\u0027ll send a mail now!","accounts_in_message":[],"_revision_number":1},{"id":"fb25f0c4ae7484c9ab634196a483e940a2632342","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-09-22 19:02:33.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review+1\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e I\u0027m tentatively okay with this, but remember that discussions which happen at the PTG aren\u0027t necessarily inclusive of the larger community, so linking to where it was discussed on the mailing list (it *was* discussed on the ML, right?) would be a good addition to the commit message for posterity.\n\u003e \u003e \u003e \n\u003e \u003e \u003e More generally, the \"unmaintained\" phase was intended to encourage teams to have a waiting period between when there was nobody to take care of backports or fix CI jobs and when the EOL tag got applied and the branch deleted. This was meant as a warning to anyone consuming that stable branch that if they didn\u0027t pitch in and help maintain it, then it would be gone permanently, since we don\u0027t re-create branches after EOL. I think getting rid of that warning period as an explicit phase is fine, but the guide could stand to recommend to projects considering EOL of an EM branch that they first provide ample warning to the broader community of the situation, in case there\u0027s someone interested in helping reverse course away from EOL before it\u0027s too late.\n\u003e \u003e \n\u003e \u003e \n\u003e \u003e This \u0027unmaintained\u0027 phase is not really used for anything since it is created. we saw stable directly going from EM to EOL like stable/ocata. I feel we do not need this as such and for warning there is a mandatory step for PTL/projects to send warning to ML before they do transition:\n\u003e \u003e \n\u003e \u003e \"To initiate this transition, either the PTL of the given project or other\n\u003e \u003e stable maintainer should:\n\u003e \u003e \n\u003e \u003e #. Send an announcement to the openstack-discuss mailing list (in order to give\n\u003e \u003e    some time for others to step up as maintainers if there are volunteers).\n\u003e \u003e .....\"\n\u003e \u003e \n\u003e \u003e \n\u003e \u003e I am ok with this but let\u0027s send it on ML also about what we agreed  TC PTG session.\n\u003e \n\u003e Ok, thank you Ghanshyam! I\u0027ll send a mail now!\n\nHere is the mail: http://lists.openstack.org/pipermail/openstack-discuss/2021-September/025016.html","accounts_in_message":[],"_revision_number":1},{"id":"f9ba010af108ce5ed4c13cfe94cb4f49e1749fb4","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-09-22 19:08:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d3e06eb350ae4cad334c66adddea795d7894aea4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-09-23 08:27:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acc0fd7124c4117e8fb5037baa2b2292248b52f4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-23 13:56:00.000000000","message":"Patch Set 1: Code-Review+2\n\nlgtm, let\u0027s wait for couple of days in case any objection on ML.\n\nThanks Elod for taking care of it.","accounts_in_message":[],"_revision_number":1},{"id":"9c7171042c8d9d00a5fa262cc9a473476b5c1edb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-23 16:33:11.000000000","message":"Patch Set 1: Code-Review+2\n\nVery for this.","accounts_in_message":[],"_revision_number":1},{"id":"bcff70517b603734fdbc289d1a56367b95568ec2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-25 15:46:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3e570ebf043fd24debda0a41402bd292881ad579","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-27 15:28:08.000000000","message":"Patch Set 1: Workflow+1\n\nlet\u0027s merge this as no objection.","accounts_in_message":[],"_revision_number":1},{"id":"1d057b0531145b861bdfebe77448df159850b886","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-09-28 21:42:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4bf8aad961fe3f655acbf0e984b185e1e59283bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 21:51:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1e8585a95244c8e9bef52e4550ed8d3 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"c3f62dc97d275ba3a6d32e8b48e858b31ef2e301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 21:51:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8281c2c5b5317f2f6bdd7be74d9af0a17bae5761","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 21:59:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67527bf6d3fc47c1a078e3b40fbd72d9 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"644cf036afdf10fb4accdb0078c8b6a4bff4d94f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 21:59:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"97ece86aba366f261c802dfd7d452fe18d393c67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 22:00:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/ad632dc55d2f44b192aac0cfb877aecb : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9c16396d656a21e9fb78e7e52162c1fc8cc81a2","revisions":{"b9c16396d656a21e9fb78e7e52162c1fc8cc81a2":{"kind":"REWORK","_number":1,"created":"2021-09-22 18:13:29.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/99/810499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/99/810499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/810499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/810499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/99/810499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/99/810499/1"}}},"commit":{"parents":[{"commit":"0a0dcda79b7d261008e1e132347eb5f5b50b4b50","subject":"Add section regarding releasing stable branches","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/0a0dcda79b7d261008e1e132347eb5f5b50b4b50"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-09-22 18:04:41.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2021-09-22 18:08:30.000000000","tz":120},"subject":"Remove \u0027Unmaintained\u0027 phase","message":"Remove \u0027Unmaintained\u0027 phase\n\nDuring Xena Virtual PTG on the Technical Committee session there was a\ndiscussion, that the \u0027Unmaintained\u0027 phase is not really clear for the\ncommunity and just creates confusion. The decision was to simply remove\nthis phase as the process is more clear and works well without it.\n\nChange-Id: I9589cbdd74d83d23ad42ce6724e3b8e4cc56c44d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/b9c16396d656a21e9fb78e7e52162c1fc8cc81a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/b9c16396d656a21e9fb78e7e52162c1fc8cc81a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
