)]}'
{"id":"openstack%2Fproject-team-guide~836392","triplet_id":"openstack%2Fproject-team-guide~master~Iadfc32f79423b54e5933788686791c42c8d616fd","project":"openstack/project-team-guide","branch":"master","topic":"rechecks","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-04-08 00:40:56.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-04-05 21:12:25.000000000","reason":"Brian Rosmaita replied on the change"},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-04-08 00:40:56.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-04-08 00:40:56.000000000","reason":"Change was submitted"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-04-08 00:40:56.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-04-05 17:17:42.000000000","reason":"Ghanshyam Mann replied on the change"},"7166":{"account":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"last_update":"2022-04-08 00:40:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iadfc32f79423b54e5933788686791c42c8d616fd","subject":"Add generic guidance on rechecks","status":"MERGED","created":"2022-04-04 18:35:40.000000000","updated":"2022-04-08 00:42:21.000000000","submitted":"2022-04-08 00:40:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"836392-rechecks","meta_rev_id":"de6380585382d46a103a2abaa305155a6ca4d50b","_number":836392,"virtual_id_number":836392,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-08 00:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2022-04-07 00:56:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2022-04-05 18:26:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":2,"date":"2022-04-05 21:12:25.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":2,"date":"2022-04-05 17:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-05 15:34:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2022-04-08 00:32:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"reviewers":{"REVIEWER":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-04 18:36:51.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-04-04 18:36:51.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2022-04-04 18:36:51.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-04-04 18:36:51.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-04-04 18:45:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-04 20:04:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2022-04-04 20:09:39.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-04-04 20:22:58.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-04-05 18:26:21.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"d4da8d05be24a2dfc49808ac0089989894b6dbf5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-04 18:35:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a6a5095a1ccda255ffeb76de81be740c0314871","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-04 18:36:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"773b7602f1612f51852272443c2d4508f54888cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 18:45:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f330c8791cc04490823d53e8f105d428 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":1},{"id":"44ad68b1360338ee8b0fc92d2b6b31a95f96ed88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-04 19:56:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"444c56255e9778bbccf6aafd08b3a82f2a4f8175","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-04-04 20:04:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94d122cebae4b3fd650809a97049f80ce10dbf54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 20:04:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03a1aa918cf74d83b653d8a84e9af72a : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"2e511063414a9760c71c674e5b5727e623ad5da7","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-04 20:09:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"838f0eacf4ed294fc35a1a5c712d2c67246f1d0a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-04 20:16:23.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"43350855cf1225c4b1347f35fa4a4dcf5ed58f37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-04 20:16:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"02245f9e5f9dce7462dcb0bb85af5a602fd8b235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 20:20:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca2a08c140a8458f9b03b4589ad37fc6 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":3},{"id":"725c2bb764406f9dbed512fe5415bf085dccd6f9","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-04-04 20:22:58.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b7efeb240c4631a1e2dd529d0dbb075fbc931dde","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-04 20:31:26.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2d8d85c15be65be0fa7b3a333fce036831ead7d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-04 20:31:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c2b9b4de4bb65d54cad872123a802590e832dd6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-04 20:35:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb7d208037cb42a9b55687b826faaf18 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":4},{"id":"f454464e23865f1fa56d743dde137fc0fc91f8f2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-04 23:19:42.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1b1472fbec2894aee18f84d9f75316627bb4daf2","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-04-05 00:11:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ca3b83b2bf3f5c2f99b666292b8d06d4571a9155","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-05 03:53:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6c8a86eb3177c55cc2f5ffb5374ef7d82e4de984","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2022-04-05 08:50:39.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6b93fca241519312285849d014cd61f9516810cf","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-05 13:03:51.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0a99884ab82aaebf9ada6bf32daf7976064a9cb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-05 13:34:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"993a0ea0dfdfef1b05f1e7171d24b305e9f015ab","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-05 13:36:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1179e2a68e1f8f9f1fdd94e8d22f79a7a8f22df3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 13:39:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0b0574c94c24cf1b9df35ea8f15c45a : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":5},{"id":"a70c15129eca555cad23cfa1f6eb2c1cd3c953fb","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-04-05 15:34:11.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"88fb8d3a21ce9ee8e625b12145a86f0fcb160781","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-05 17:17:42.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6ba9d42487502ccd735068168c740400e131cb42","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-05 17:36:51.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"6a1c24b3d4fe0021f1c45031d373d95850ee5b87","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2022-04-05 18:26:21.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7c8b47b1761ab3e5b2162b555812699f73e1429e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-05 19:21:47.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7341e6a3d20a58f1f1d6423cb912fbb1c7f15d82","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-04-05 19:51:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"82b4fe49d5966aa3b7aebaf976752f3133e4fb92","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-04-05 19:54:42.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a3d2f768e54ae5cf8433ec1725148af841dc7ce4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-04-05 21:12:25.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e9b33b6639f5119eb9cbcf59e2e229d913559c4f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2022-04-07 00:56:49.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"31b7b3f4716a180734eb124ebe4f81f2dc94c5af","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-04-08 00:32:34.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"95ee394b8846e33e061cc8a1501e738699957448","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 00:32:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"564549a1ca97de0f9191096914b4e6783012d58d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 00:40:56.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b57e24e651014dbcbed5e6851e6a1674 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":5},{"id":"606a0ca36dc9397fcf162b337f2bca759b249c06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 00:40:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"de6380585382d46a103a2abaa305155a6ca4d50b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 00:42:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/4849e16d068944f49b6182f510515ef1 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ef313d1f8eb4bccbf91e8621a417651602444972","revisions":{"6d9bac895ef1e0b3f655616d6d80cf067452f4f6":{"kind":"REWORK","_number":1,"created":"2022-04-04 18:35:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/836392/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/92/836392/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/1"}}},"commit":{"parents":[{"commit":"f4f4325a0ad99fb74d4d5c6237aac57ba911255f","subject":"Merge \"Update note about periodic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/f4f4325a0ad99fb74d4d5c6237aac57ba911255f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-04-04 19:40:09.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-04-04 19:43:33.000000000","tz":-420},"subject":"Add generic guidance on rechecks","message":"Add generic guidance on rechecks\n\nPer recent discussions, this attempts to provide some general\nguidance about responsible use of rechecks that we can link to when\n\"encouraging\" people to exhibit the desired behavior.\n\nChange-Id: Iadfc32f79423b54e5933788686791c42c8d616fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/6d9bac895ef1e0b3f655616d6d80cf067452f4f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/6d9bac895ef1e0b3f655616d6d80cf067452f4f6"}]},"branch":"refs/heads/master"},"3f09f363447be7cdaeed7d983bf28ebf5ba32707":{"kind":"REWORK","_number":2,"created":"2022-04-04 19:56:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/836392/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/92/836392/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/2"}}},"commit":{"parents":[{"commit":"f4f4325a0ad99fb74d4d5c6237aac57ba911255f","subject":"Merge \"Update note about periodic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/f4f4325a0ad99fb74d4d5c6237aac57ba911255f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-04-04 19:40:09.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-04-04 21:04:30.000000000","tz":-420},"subject":"Add generic guidance on rechecks","message":"Add generic guidance on rechecks\n\nPer recent discussions, this attempts to provide some general\nguidance about responsible use of rechecks that we can link to when\n\"encouraging\" people to exhibit the desired behavior.\n\nChange-Id: Iadfc32f79423b54e5933788686791c42c8d616fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/3f09f363447be7cdaeed7d983bf28ebf5ba32707"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/3f09f363447be7cdaeed7d983bf28ebf5ba32707"}]},"branch":"refs/heads/master"},"b11b031613c83db689d52e4971c8c568fa3da7af":{"kind":"REWORK","_number":3,"created":"2022-04-04 20:16:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/836392/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/92/836392/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/3"}}},"commit":{"parents":[{"commit":"f4f4325a0ad99fb74d4d5c6237aac57ba911255f","subject":"Merge \"Update note about periodic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/f4f4325a0ad99fb74d4d5c6237aac57ba911255f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-04-04 19:40:09.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-04-04 21:24:29.000000000","tz":-420},"subject":"Add generic guidance on rechecks","message":"Add generic guidance on rechecks\n\nPer recent discussions, this attempts to provide some general\nguidance about responsible use of rechecks that we can link to when\n\"encouraging\" people to exhibit the desired behavior.\n\nChange-Id: Iadfc32f79423b54e5933788686791c42c8d616fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/b11b031613c83db689d52e4971c8c568fa3da7af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/b11b031613c83db689d52e4971c8c568fa3da7af"}]},"branch":"refs/heads/master"},"821860cb4d8e4650a5228070d42c2091ed249141":{"kind":"REWORK","_number":4,"created":"2022-04-04 20:31:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/836392/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/92/836392/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/4"}}},"commit":{"parents":[{"commit":"f4f4325a0ad99fb74d4d5c6237aac57ba911255f","subject":"Merge \"Update note about periodic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/f4f4325a0ad99fb74d4d5c6237aac57ba911255f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-04-04 19:40:09.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-04-04 21:39:29.000000000","tz":-420},"subject":"Add generic guidance on rechecks","message":"Add generic guidance on rechecks\n\nPer recent discussions, this attempts to provide some general\nguidance about responsible use of rechecks that we can link to when\n\"encouraging\" people to exhibit the desired behavior.\n\nChange-Id: Iadfc32f79423b54e5933788686791c42c8d616fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/821860cb4d8e4650a5228070d42c2091ed249141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/821860cb4d8e4650a5228070d42c2091ed249141"}]},"branch":"refs/heads/master"},"ef313d1f8eb4bccbf91e8621a417651602444972":{"kind":"REWORK","_number":5,"created":"2022-04-05 13:34:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/92/836392/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/92/836392/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/92/836392/5"}}},"commit":{"parents":[{"commit":"f4f4325a0ad99fb74d4d5c6237aac57ba911255f","subject":"Merge \"Update note about periodic jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/f4f4325a0ad99fb74d4d5c6237aac57ba911255f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-04-04 19:40:09.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-04-05 14:42:32.000000000","tz":-420},"subject":"Add generic guidance on rechecks","message":"Add generic guidance on rechecks\n\nPer recent discussions, this attempts to provide some general\nguidance about responsible use of rechecks that we can link to when\n\"encouraging\" people to exhibit the desired behavior.\n\nChange-Id: Iadfc32f79423b54e5933788686791c42c8d616fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ef313d1f8eb4bccbf91e8621a417651602444972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/ef313d1f8eb4bccbf91e8621a417651602444972"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
