)]}'
{"id":"openstack%2Fproject-team-guide~894610","triplet_id":"openstack%2Fproject-team-guide~master~Idd16e92c2ddd0b3571275dd52777422a2af320fd","project":"openstack/project-team-guide","branch":"master","attention_set":{},"removed_from_attention_set":{"11904":{"account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"last_update":"2023-09-13 11:33:50.000000000","reason":"\u003cGERRIT_ACCOUNT_11904\u003e replied on the change","reason_account":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},"35025":{"account":{"_account_id":35025,"deleted":true},"last_update":"2023-09-13 11:38:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idd16e92c2ddd0b3571275dd52777422a2af320fd","subject":"Add and remove incorrect links","status":"MERGED","created":"2023-09-12 03:18:39.000000000","updated":"2023-09-13 11:39:09.000000000","submitted":"2023-09-13 11:38:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894610","meta_rev_id":"028046c5775b36bf777c9c24a42a1eb4c38e9c00","_number":894610,"virtual_id_number":894610,"owner":{"_account_id":35025,"deleted":true},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":35025},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-13 11:38:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2023-09-13 11:33:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2023-09-13 07:55:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35025},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2023-09-13 11:33:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":35025},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35025,"deleted":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-12 03:27:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-12 11:30:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2023-09-13 11:33:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"abed34cc7c6779539375ed919f5d14da6986566b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2023-09-12 03:18:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"054fd6d657b3def62b6ce77f1bc12eae260ca81a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 03:27:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1bfbfeadd3a46afa99e742470bf9688\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcc34f2abbea43c187b1f49076adc2af : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"487146cc7c1cb58062b65feaef3f5c8881291743","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-09-12 11:30:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db0cdac363e0b5000b715a18f29403cc59617c05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35025,"deleted":true},"date":"2023-09-13 01:51:35.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"902ac0c29766ad4c38aef49fe3a1c0af008f69e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 02:02:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcabf921a37c4c5e9efb2d16ae761c19\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccadeba467444d3585c08b8dcbedbef2 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3cbf64284bfb115b8565abab033855e1d9d8e9f6","author":{"_account_id":35025,"deleted":true},"date":"2023-09-13 07:55:22.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f3ef0247d5edd12717d35e667ee7414edbdde0f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2023-09-13 11:33:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"87510b1a914d3c410e25316410d87faec2a5c2af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 11:34:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c34e688c39aadfe0008d266278ae2f38839e36df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 11:38:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83d875c0ad1c4fa887f42827715e531e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed4a39dc507343d1b17bffe707ef314d : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"558ee93f746db1db76ab59bbbf59531138fda9ae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 11:38:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"028046c5775b36bf777c9c24a42a1eb4c38e9c00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 11:39:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe40b9a7ee70497f8cc85a52171b9868\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/7035c4d92a824958917986358e0984c0 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0bc374d381200830da6907599987ac3497a36b14","revisions":{"fb51757a215e3c36dc29a0cb7aab8b0adc346675":{"kind":"REWORK","_number":1,"created":"2023-09-12 03:18:39.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/10/894610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/10/894610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/10/894610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/10/894610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/10/894610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/10/894610/1"}}},"commit":{"parents":[{"commit":"fa6f40684c8fecbfa23eb638311e43317633ad5a","subject":"Add SLURP release notes strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/fa6f40684c8fecbfa23eb638311e43317633ad5a"}]}],"author":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-09-12 03:17:01.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-09-12 03:18:36.000000000","tz":480},"subject":"Add and remove incorrect links","message":"Add and remove incorrect links\n\nChange-Id: Idd16e92c2ddd0b3571275dd52777422a2af320fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/fb51757a215e3c36dc29a0cb7aab8b0adc346675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/fb51757a215e3c36dc29a0cb7aab8b0adc346675"}]},"branch":"refs/heads/master"},"0bc374d381200830da6907599987ac3497a36b14":{"kind":"REWORK","_number":2,"created":"2023-09-13 01:51:35.000000000","uploader":{"_account_id":35025,"deleted":true},"ref":"refs/changes/10/894610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/10/894610/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/10/894610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/10/894610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/10/894610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/10/894610/2"}}},"commit":{"parents":[{"commit":"fa6f40684c8fecbfa23eb638311e43317633ad5a","subject":"Add SLURP release notes strategy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/fa6f40684c8fecbfa23eb638311e43317633ad5a"}]}],"author":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-09-12 03:17:01.000000000","tz":480},"committer":{"name":"renliang","email":"renliang@uniontech.com","date":"2023-09-13 01:51:21.000000000","tz":480},"subject":"Add and remove incorrect links","message":"Add and remove incorrect links\n\nChange-Id: Idd16e92c2ddd0b3571275dd52777422a2af320fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/0bc374d381200830da6907599987ac3497a36b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/0bc374d381200830da6907599987ac3497a36b14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
