)]}'
{"id":"openstack%2Fproject-team-guide~911375","triplet_id":"openstack%2Fproject-team-guide~master~Ica33e5cfe0e5ba84e1bf648393eb2c7b788be5f1","project":"openstack/project-team-guide","branch":"master","topic":"transition","attention_set":{},"removed_from_attention_set":{"19298":{"account":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"last_update":"2024-03-05 22:50:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ica33e5cfe0e5ba84e1bf648393eb2c7b788be5f1","subject":"Add section about transition","status":"ABANDONED","created":"2024-03-05 20:25:41.000000000","updated":"2024-03-05 22:50:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f04dcb2b17851eb6d9388f1a926e35ca6f6d3ca0","_number":911375,"virtual_id_number":911375,"owner":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-05 20:29:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-05 20:29:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba1b80ed0d547595634f61aacf49180db7c51296","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2024-03-05 20:25:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27b3313efdddc5b3348c270be4bc4b6c7fcdc03a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 20:29:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7599f53a36bc4026a9e39bc2a18a6f39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af21fe2e59e84d2fbd2502dfd3e282a8 : FAILURE in 2m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f04dcb2b17851eb6d9388f1a926e35ca6f6d3ca0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2024-03-05 22:50:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3591921da0b6e1ae753b4022a15939f7d56778f3","revisions":{"3591921da0b6e1ae753b4022a15939f7d56778f3":{"kind":"REWORK","_number":1,"created":"2024-03-05 20:25:41.000000000","uploader":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"ref":"refs/changes/75/911375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/75/911375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/75/911375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/75/911375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/75/911375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/75/911375/1"}}},"commit":{"parents":[{"commit":"7cc5e309c32b4ce81d41f83077e28b6dd943b5a7","subject":"Merge \"Update docs for Unmaintained\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/7cc5e309c32b4ce81d41f83077e28b6dd943b5a7"}]}],"author":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2024-03-05 20:24:22.000000000","tz":-420},"committer":{"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","date":"2024-03-05 20:25:40.000000000","tz":-420},"subject":"Add section about transition","message":"Add section about transition\n\nThe `unmaintained` section does not clearly spell out how the transition\nto this new branch structure is to be transitioned for branches before\nthe first SLURP release.\n\nChange-Id: Ica33e5cfe0e5ba84e1bf648393eb2c7b788be5f1\nSigned-off-by: Nicolas Bock \u003cnicolas.bock@canonical.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/3591921da0b6e1ae753b4022a15939f7d56778f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/3591921da0b6e1ae753b4022a15939f7d56778f3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
