)]}'
{"id":"openstack%2Fproject-team-guide~919976","triplet_id":"openstack%2Fproject-team-guide~master~I9acad0e1a64ad34ae85314f6cf355d3bfaf9fc82","project":"openstack/project-team-guide","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-05-17 19:53:00.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-05-21 18:40:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9acad0e1a64ad34ae85314f6cf355d3bfaf9fc82","subject":"Make the project removal from infra as step#6","status":"MERGED","created":"2024-05-17 17:28:16.000000000","updated":"2024-05-21 18:41:50.000000000","submitted":"2024-05-21 18:40:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919976","meta_rev_id":"051584fc519811bd2f63c4914b2185a15815275f","_number":919976,"virtual_id_number":919976,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-21 18:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2024-05-17 19:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-05-21 18:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2024-05-21 18:37:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-17 17:32:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-17 19:09:37.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-05-21 18:37:19.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"847e181d3b540a79d70593e01c8603a904003601","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-17 17:28:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fcd0190c95e581e735be3b2675655e9f8243aa0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-17 17:28:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d7b26f4599dd7a00ef9206b60a1c34f35319c600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 17:32:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f1847be3d8b417aa03ea1e097f06f3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb11be25589d4cc4a9efb7d13ae308b2 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":2},{"id":"6eabe490bdc567c917fbf82a79630cb7fe33d596","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-17 19:09:37.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"0011facc96c2dbb15c61c6aac4b37b55c2443464","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-17 19:49:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f408c7be72f9dc93b038ff4606554c9ed8be1c77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-05-17 19:50:34.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9b1c987be3714036fc91e93782c41fc8dbe5d29a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-17 19:53:00.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f36e3aaeb867ce73c9c292c2fa0b0db86c7812dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 19:54:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39542a953619410c94a7f21f0d836c3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d48c724b8343a8b74a5ab7ae151a89 : SUCCESS in 2m 37s","accounts_in_message":[],"_revision_number":3},{"id":"7dd5fb32f8644ddf70ae84154a2baec0c52cfb98","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-05-21 18:37:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f6e3f528ce521beaffdc306206445eda4eff85cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 18:37:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b2e654550b931f3fc0cd4721b44aac1a08dacaf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 18:40:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43693f57cdc0449ab757284d57ba5374\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac7c0f99aa91454ba8506db55a1a2998 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":3},{"id":"ecb2e4490106846e51f13fb2ff3f09e12ce55e50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 18:40:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"051584fc519811bd2f63c4914b2185a15815275f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 18:41:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3db012e22fb84393a2caaa97bd38e5da\n\n- promote-openstack-tox-docs-direct https://zuul.opendev.org/t/openstack/build/00a0525c526a4e1a904177a571d25f34 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0166f33a5dfadd39d38127192713e8b73cc4ee7f","revisions":{"046a75904de6cd6669cbf21ea5454c997291ae54":{"kind":"REWORK","_number":1,"created":"2024-05-17 17:28:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/76/919976/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/76/919976/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/1"}}},"commit":{"parents":[{"commit":"5338fd07f4d631d23ddbb00f79cf850e61552529","subject":"Improve terminology in project-team-guide tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/5338fd07f4d631d23ddbb00f79cf850e61552529"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-17 17:24:05.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-17 17:24:05.000000000","tz":-420},"subject":"Make the project removal from infra as step#6","message":"Make the project removal from infra as step#6\n\nWhen project is being removed from infra then it will\nstart giving error whereever that project is used in zuul\njobs (as required-projects).\n\nTo avoid such error, we can cleanup the reference of retired\nproject first and then remove it from infra.\n\nML Discussion: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/24M3WQXDE2DYN3GWDFHY52FNGNIXN4N3/\n\nChange-Id: I9acad0e1a64ad34ae85314f6cf355d3bfaf9fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/046a75904de6cd6669cbf21ea5454c997291ae54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/046a75904de6cd6669cbf21ea5454c997291ae54"}]},"branch":"refs/heads/master"},"817da94526677610bfc52c7f7439ba1701f7a0df":{"kind":"REWORK","_number":2,"created":"2024-05-17 17:28:49.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/76/919976/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/76/919976/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/2"}}},"commit":{"parents":[{"commit":"5338fd07f4d631d23ddbb00f79cf850e61552529","subject":"Improve terminology in project-team-guide tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/5338fd07f4d631d23ddbb00f79cf850e61552529"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-17 17:24:05.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-17 17:28:43.000000000","tz":-420},"subject":"Make the project removal from infra as step#6","message":"Make the project removal from infra as step#6\n\nWhen project is being removed from infra then it will\nstart giving error whereever that project is used in zuul\njobs (as required-projects).\n\nTo avoid such error, we can cleanup the reference of retired\nproject first and then remove it from infra.\n\nML Discussion: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/24M3WQXDE2DYN3GWDFHY52FNGNIXN4N3/\n\nChange-Id: I9acad0e1a64ad34ae85314f6cf355d3bfaf9fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/817da94526677610bfc52c7f7439ba1701f7a0df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/817da94526677610bfc52c7f7439ba1701f7a0df"}]},"branch":"refs/heads/master"},"0166f33a5dfadd39d38127192713e8b73cc4ee7f":{"kind":"REWORK","_number":3,"created":"2024-05-17 19:50:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/76/919976/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/project-team-guide","ref":"refs/changes/76/919976/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/project-team-guide refs/changes/76/919976/3"}}},"commit":{"parents":[{"commit":"5338fd07f4d631d23ddbb00f79cf850e61552529","subject":"Improve terminology in project-team-guide tree","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/5338fd07f4d631d23ddbb00f79cf850e61552529"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-05-17 17:24:05.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-05-17 19:50:33.000000000","tz":0},"subject":"Make the project removal from infra as step#6","message":"Make the project removal from infra as step#6\n\nWhen project is being removed from infra then it will\nstart giving error whereever that project is used in zuul\njobs (as required-projects).\n\nTo avoid such error, we can cleanup the reference of retired\nproject first and then remove it from infra.\n\nML Discussion: https://lists.openstack.org/archives/list/openstack-discuss@lists.openstack.org/thread/24M3WQXDE2DYN3GWDFHY52FNGNIXN4N3/\n\nChange-Id: I9acad0e1a64ad34ae85314f6cf355d3bfaf9fc82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/0166f33a5dfadd39d38127192713e8b73cc4ee7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/project-team-guide/commit/0166f33a5dfadd39d38127192713e8b73cc4ee7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
