)]}'
{"id":"openstack%2Fptgbot~656110","triplet_id":"openstack%2Fptgbot~master~I1b86c1f403981b066ad096548a1754fadd6b72ad","project":"openstack/ptgbot","branch":"master","topic":"presence-tracking","hashtags":[],"change_id":"I1b86c1f403981b066ad096548a1754fadd6b72ad","subject":"Show check-ins per room in \"Currently at the PTG\" page","status":"MERGED","created":"2019-04-28 05:31:08.000000000","updated":"2019-04-30 21:16:25.000000000","submitted":"2019-04-30 21:16:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656110-1556658985522-82ae42cf","meta_rev_id":"8d75d36c4ec460b784d4dbe68f7bec41833eebb6","_number":656110,"virtual_id_number":656110,"owner":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-04-28 17:33:18.000000000","_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":2,"date":"2019-04-30 21:16:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-04-28 17:45:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-04-28 20:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-04-28 20:46:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":2,"date":"2019-04-30 21:10:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-29 00:06:25.000000000","_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"value":1,"date":"2019-04-30 21:10:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-28 17:45:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-04-28 20:11:43.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-04-28 20:46:09.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-04-29 00:06:25.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2019-04-30 21:10:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-04-30 21:16:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e415dea43c024b92ba8005dfa62fa3f3aa8f1644","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-28 05:31:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4027a3cb96c1c500e4055ad3f41a996e930575d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 05:35:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/656110/1/check/openstack-tox-pep8/a8e6d5f/ : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"243951d7a74817321dff0820ee6292c27af54a05","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-28 12:39:01.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nNeat","accounts_in_message":[],"_revision_number":1},{"id":"e4618ce9d3efdb4110c59d52e319dfef8841c6d0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-28 17:12:05.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good to me. Typo fix would be good, but I would not be too concerned if it went through as it is.","accounts_in_message":[],"_revision_number":1},{"id":"fad682f3d9ccfaf25b45b6c8d4ef9c64bd0c8c9d","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-28 17:32:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"47999660df69a4eb740c88a67a585b4ffe954e25","author":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"date":"2019-04-28 17:33:18.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nAlso spotted and in patch set 2 fixed a minor bug when everyone who had checked into a room had subsequently checked out.","accounts_in_message":[],"_revision_number":2},{"id":"022861c4965ee18ff09c2107177851a019791b46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 17:40:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/656110/2/check/openstack-tox-pep8/8e22f44/ : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":2},{"id":"88177fd7bcd77744da23af5f6d33b10e18c00898","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-04-28 17:45:39.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4baa2eb78930297030965fe3fd23e4814305bd0c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-04-28 20:11:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"97163bfafe260276aea7722f70382ce5260ece19","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-28 20:46:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"35df9231246a02d78298fd47db7eb324ebd230de","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2019-04-30 21:10:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b709d9f399c22ca715f2d76624be12d3095b67dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 21:11:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7edd83ca512816a33e713998d0b54d1743c59e1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 21:16:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0e8bbe684c5900828f443c9d4399af5dd17ccbfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 21:16:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/10/656110/2/gate/openstack-tox-pep8/89c3589/ : SUCCESS in 4m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a495d3011b8ba0067be337b6dc43e33c5488fd41","revisions":{"e48b1b6f4964e66499e486030bc5581a908de2a4":{"kind":"REWORK","_number":1,"created":"2019-04-28 05:31:08.000000000","uploader":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"ref":"refs/changes/10/656110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/10/656110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/10/656110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/10/656110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/10/656110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/10/656110/1"}}},"commit":{"parents":[{"commit":"946d69b140c89dc93141a8e68696cecf29c009dc","subject":"Add presence tracking commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/946d69b140c89dc93141a8e68696cecf29c009dc"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-28 00:48:05.000000000","tz":-360},"committer":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-28 00:48:05.000000000","tz":-360},"subject":"Show check-ins per room in \"Currently at the PTG\" page","message":"Show check-ins per room in \"Currently at the PTG\" page\n\nInclude tooltips on the \"Currently at the PTG\" page, so that hovering\nthe mouse over a particular track will show IRC nicks for everyone\ncurrently checked into that track.\n\nChange-Id: I1b86c1f403981b066ad096548a1754fadd6b72ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/e48b1b6f4964e66499e486030bc5581a908de2a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/e48b1b6f4964e66499e486030bc5581a908de2a4"}]},"branch":"refs/heads/master"},"a495d3011b8ba0067be337b6dc43e33c5488fd41":{"kind":"REWORK","_number":2,"created":"2019-04-28 17:32:08.000000000","uploader":{"_account_id":2394,"name":"Adam Spiers","email":"aspiers@suse.com","username":"adam.spiers"},"ref":"refs/changes/10/656110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/10/656110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/10/656110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/10/656110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/10/656110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/10/656110/2"}}},"commit":{"parents":[{"commit":"946d69b140c89dc93141a8e68696cecf29c009dc","subject":"Add presence tracking commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/946d69b140c89dc93141a8e68696cecf29c009dc"}]}],"author":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-28 00:48:05.000000000","tz":-360},"committer":{"name":"Adam Spiers","email":"aspiers@suse.com","date":"2019-04-28 17:30:52.000000000","tz":-360},"subject":"Show check-ins per room in \"Currently at the PTG\" page","message":"Show check-ins per room in \"Currently at the PTG\" page\n\nInclude tooltips on the \"Currently at the PTG\" page, so that hovering\nthe mouse over a particular track will show IRC nicks for everyone\ncurrently checked into that track.  If noone has checked in to the\ntrack yet, give a hint about how to do it.\n\nThanks a lot to Jay Bryant for giving me the idea for this after we\nbumped into each other on the plane to Denver ;-)\n\nChange-Id: I1b86c1f403981b066ad096548a1754fadd6b72ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/a495d3011b8ba0067be337b6dc43e33c5488fd41"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/a495d3011b8ba0067be337b6dc43e33c5488fd41"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
