)]}'
{"id":"openstack%2Fptgbot~674606","triplet_id":"openstack%2Fptgbot~master~I343c28acc3ad6015d3382d526511f153bf38a0e6","project":"openstack/ptgbot","branch":"master","topic":"room-capabilities","hashtags":[],"change_id":"I343c28acc3ad6015d3382d526511f153bf38a0e6","subject":"Display room capabilities","status":"MERGED","created":"2019-08-05 13:38:21.000000000","updated":"2019-08-23 15:16:14.000000000","submitted":"2019-08-23 15:16:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674606-1566573374821-38c415cc","meta_rev_id":"2b8627f1e1abd2facfe73c20457ce8d0707fd7ad","_number":674606,"virtual_id_number":674606,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-08-23 15:16:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-08-23 15:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-08-23 15:08:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-23 15:16:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e582aa7ccf597e52ed31d06ba1d0d0b56cf5af97","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-05 13:38:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edaa77291499ed55c100738637ce6c9e20dbda70","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:46:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/06/674606/1/check/openstack-tox-pep8/d768055/ : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ea11c6c2c3313467f32f15df9b5b425589bf6997","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-05 14:52:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a335ad88359e5f29bcc75ce4dc0f23468f540b43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 15:01:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/06/674606/2/check/openstack-tox-pep8/043ce41/ : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":2},{"id":"31dabb02ccf5353b57e47acc13ae9230b6749b50","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-08-23 15:08:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"457984d32738b86e64b6d5409684fb417777c68a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:08:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6b295f9dbb7b5f298a8ece98f9b340dbbe49dd7f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:16:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1a5bb85c0e1cbcd677e4d71a9e9e4122b9aa79c0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 15:16:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dcdbc5b4451468c873958224e25eb2c : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e1f5c3715b5c4934565396d89f7c3c56fce2bc2","revisions":{"b5e50c474a465b8ae7b341a4d56f8430cb4fa916":{"kind":"REWORK","_number":1,"created":"2019-08-05 13:38:21.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/06/674606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/06/674606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/06/674606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/06/674606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/06/674606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/06/674606/1"}}},"commit":{"parents":[{"commit":"651a48c8c11edf6ada20f84074388f043db7401e","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/651a48c8c11edf6ada20f84074388f043db7401e"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-05 12:49:18.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-05 13:36:31.000000000","tz":120},"subject":"Display room capabilities","message":"Display room capabilities\n\nIt is sometimes useful to mark certain rooms as having specific\ncapabilities like videoprojection. This change adds the possibility\nto assign a glyphicon (cap_icon) and a mouseover description\n(cap_desc) for the rooms in the schedule dictionary.\n\nChange-Id: I343c28acc3ad6015d3382d526511f153bf38a0e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/b5e50c474a465b8ae7b341a4d56f8430cb4fa916"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/b5e50c474a465b8ae7b341a4d56f8430cb4fa916"}]},"branch":"refs/heads/master"},"7e1f5c3715b5c4934565396d89f7c3c56fce2bc2":{"kind":"REWORK","_number":2,"created":"2019-08-05 14:52:05.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/06/674606/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/06/674606/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/06/674606/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/06/674606/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/06/674606/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/06/674606/2"}}},"commit":{"parents":[{"commit":"651a48c8c11edf6ada20f84074388f043db7401e","subject":"Merge \"Add Python 3 Train unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/651a48c8c11edf6ada20f84074388f043db7401e"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-05 12:49:18.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-08-05 14:51:10.000000000","tz":120},"subject":"Display room capabilities","message":"Display room capabilities\n\nIt is sometimes useful to mark certain rooms as having specific\ncapabilities like videoprojection. This change adds the possibility\nto assign a glyphicon (cap_icon) and a mouseover description\n(cap_desc) for the rooms in the schedule dictionary.\n\nChange-Id: I343c28acc3ad6015d3382d526511f153bf38a0e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/7e1f5c3715b5c4934565396d89f7c3c56fce2bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/7e1f5c3715b5c4934565396d89f7c3c56fce2bc2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
