)]}'
{"id":"openstack%2Fptgbot~725249","triplet_id":"openstack%2Fptgbot~master~I633eb83c579e4093ae769bcd053e4d2ec2fe4bc3","project":"openstack/ptgbot","branch":"master","topic":"virtual-ptg-support","hashtags":[],"change_id":"I633eb83c579e4093ae769bcd053e4d2ec2fe4bc3","subject":"Support for room and track URLs","status":"MERGED","created":"2020-05-04 12:13:23.000000000","updated":"2020-05-11 16:04:59.000000000","submitted":"2020-05-11 16:04:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725249-1589213099372-0a89928d","meta_rev_id":"9db3f3cb360b3e18e8e289ee77cbf8977b222cae","_number":725249,"virtual_id_number":725249,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-11 15:50:07.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-11 16:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-11 13:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-11 13:13:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 13:13:36.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-05-11 16:04:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bff7c5c732e5a80a30e2a658ba225816a069be4","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-04 12:13:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"13660d5be06c80d5aaac6e67f443cdf78b504efa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 12:18:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b68cd498354bd99146d29fc1b8c05f : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":1},{"id":"0643b6d8ce64d8d76a589e357c4faaf6f6377d4d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-05 10:18:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"82535ec01d0fc329eac252be0831c140882dbfd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 10:24:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a2f5ef52cc046ce947f88aea9f14713 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":2},{"id":"a8aca1622f57da87ce35323863b4606eec359f64","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-05-11 13:13:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cb97566a4b4776719d56c2c70911bd5241259413","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 13:13:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"89b46147711ffca4c84be5f856f789e2f22fc3ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 13:16:55.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"3a1bdebeac2e51868c0141915d585e747ca78fcc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-05-11 15:50:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"068eda9948bf332112ac9c2e5a2113e826d6def2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 15:59:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/705af5773d0b402ca13faa00cd9e7ebe : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5525af9010b26ef7903cfee05e27c3ecea3a6405","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 15:59:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0b666dbd1371ace31a446eb15ad9dd9a12b0c8f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:04:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/831dcceaba0044249977b0d6058f3245 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"f67665b901bc7aac4b3c09d984e79e12c2584e51","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:04:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"553812776834127dd62f515882f0b85e20e8c834","revisions":{"33bf2e7f62c70877594cdeddc615b5938dbf52cf":{"kind":"REWORK","_number":1,"created":"2020-05-04 12:13:23.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/49/725249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/49/725249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/49/725249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/49/725249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/49/725249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/49/725249/1"}}},"commit":{"parents":[{"commit":"d2d7d21e98de873eaa1973b1b6c819e7b65b7068","subject":"Add base.json as example base data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/d2d7d21e98de873eaa1973b1b6c819e7b65b7068"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-05-04 12:09:31.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-05-04 12:09:31.000000000","tz":120},"subject":"Support for room and track URLs","message":"Support for room and track URLs\n\nAllow to associate a URL to a room, for example a link to a video\nconference room. Tracks assigned to this room will automatically inherit\nthe link. The link for a given track can be overridden using the \"url\"\ncommand.\n\nThis allows organizers to provide video links for each room, while still\nenabling teams to link to whatever they end up really using.\n\nChange-Id: I633eb83c579e4093ae769bcd053e4d2ec2fe4bc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/33bf2e7f62c70877594cdeddc615b5938dbf52cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/33bf2e7f62c70877594cdeddc615b5938dbf52cf"}]},"branch":"refs/heads/master"},"553812776834127dd62f515882f0b85e20e8c834":{"kind":"REWORK","_number":2,"created":"2020-05-05 10:18:04.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/49/725249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/49/725249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/49/725249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/49/725249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/49/725249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/49/725249/2"}}},"commit":{"parents":[{"commit":"d2d7d21e98de873eaa1973b1b6c819e7b65b7068","subject":"Add base.json as example base data","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/d2d7d21e98de873eaa1973b1b6c819e7b65b7068"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-05-04 12:09:31.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-05-05 10:17:51.000000000","tz":120},"subject":"Support for room and track URLs","message":"Support for room and track URLs\n\nAllow to associate a URL to a room, for example a link to a video\nconference room. Tracks assigned to this room will automatically inherit\nthe link. The link for a given track can be overridden using the \"url\"\ncommand.\n\nThis allows organizers to provide video links for each room, while still\nenabling teams to link to whatever they end up really using.\n\nChange-Id: I633eb83c579e4093ae769bcd053e4d2ec2fe4bc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/553812776834127dd62f515882f0b85e20e8c834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/553812776834127dd62f515882f0b85e20e8c834"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
