)]}'
{"id":"openstack%2Fptgbot~764363","triplet_id":"openstack%2Fptgbot~master~I15313f93e728a692820ac79644d70928d87b3f2a","project":"openstack/ptgbot","branch":"master","topic":"refactor-cmds","hashtags":[],"change_id":"I15313f93e728a692820ac79644d70928d87b3f2a","subject":"Add tests for user commands","status":"MERGED","created":"2020-11-26 15:51:59.000000000","updated":"2020-12-08 15:50:54.000000000","submitted":"2020-12-08 15:50:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764363-refactor-cmds","meta_rev_id":"e225d26ad05c7fd813271f97c66aaf1cc7d6ca12","_number":764363,"virtual_id_number":764363,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-08 15:50:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-08 15:39:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-08 15:39:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-26 16:05:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-08 00:45:36.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"e7a8da9532f4d4b23a9d74de2984d09c2012ee59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-11-26 15:51:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73d6e1f730101b10fd5b2b05a0e1949e82d0c5d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 16:05:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53060cee3fb249658e839ebf81b6a133 : SUCCESS in 4m 59s\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/2c142cccd7084ccca1038ec2ba96c8a1 : SUCCESS in 5m 18s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/644550e18aec4067b22280a45943785a : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ab578393b3891efdffadf3c58f22deb717cc9e44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-11-26 16:42:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ecec0099d8ca08e456ab269b827c9b74860e44f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 16:52:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/987f369e6fcb446c841f89c9729f18f6 : SUCCESS in 2m 56s\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/68133ae08fed45a7b8a212fd138e5598 : SUCCESS in 3m 51s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/12a359b1c82f4122a50ddcf9900f51b8 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f3f89e3cdc44ec6427c73bb5c5486b143921da37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-11-27 08:50:51.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"294f06ad57d0fa37ff283d1984a99f68c0f1a459","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 09:03:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee867f7975014901afbf8d3fc41720cc : SUCCESS in 2m 42s\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/20029cc1aec64080a1b6d4617b7e594f : SUCCESS in 2m 34s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/5103998723bb4ce7b7498d6be5200d8a : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b8ae92895d14b2c703bd10cfce735c5bcdfb1c70","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-12-08 00:45:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6a111f77f13933cd5c2fb8c40ad921615ef95abd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 00:45:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f80aadbf7542e8c38bddeda9e015e4d0c451d4d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 00:49:37.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6f81c89a5f24d1386cdb34503a2fc2f : SUCCESS in 2m 18s\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/8d3e7a19588048d1afff6a6f22e29cf8 : SUCCESS in 2m 14s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/f4f59c47bd4448a59d49640b0e33ffd5 : FAILURE in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"27cfb05e6645cd9e6bff9b32ed3ba31b9d7bad5a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-12-08 13:29:59.000000000","message":"Patch Set 3:\n\nAh, that is funny. This test only passes on Wednesdays, Thursdays and Fridays. as it answers a different message on days the Nova track is not scheduled 😊","accounts_in_message":[],"_revision_number":3},{"id":"7df0ccddb1e50cca7894224c3934e09ece40beaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-12-08 13:36:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba8749e588f62a8642bb4b68f1891dfb3064ffd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 13:45:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd63a4b8ce0b4d88844d23cc64eeeae0 : SUCCESS in 2m 41s\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/33a1371d5f3e4bd8882e503f9502daa1 : SUCCESS in 2m 33s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/b921645845aa4e97be48905b658944bf : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":4},{"id":"ad236c394c2c0b52af88dcf049ce5b90620bff20","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-12-08 15:39:33.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nBwahahahaha!","accounts_in_message":[],"_revision_number":4},{"id":"9c4ce7c8abead541c9035e72766560fe3fe3a57d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 15:39:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"d547fd6184b59d253fbf297571692fd80f5ff491","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 15:50:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e3b89c67154fdca8b526dbca8a6619 : SUCCESS in 2m 35s\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/658bf31058cc43daba0c08ed2cd1bddb : SUCCESS in 3m 00s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/fe2d2d9978954e80815f9c618ad674e3 : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":4},{"id":"e225d26ad05c7fd813271f97c66aaf1cc7d6ca12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 15:50:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f368b40e98513362d305b592d67f358e5a452ef6","revisions":{"bf89dfbd788c6a1ac69d22edd3069284972fd34e":{"kind":"REWORK","_number":1,"created":"2020-11-26 15:51:59.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/63/764363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/63/764363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/1"}}},"commit":{"parents":[{"commit":"8cc5cdd3a3ebc5559d94f567e90a2ceb1a3d0e54","subject":"Add tests for PTGbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/8cc5cdd3a3ebc5559d94f567e90a2ceb1a3d0e54"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-26 15:50:58.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-26 15:50:58.000000000","tz":60},"subject":"Add tests for user commands","message":"Add tests for user commands\n\nAdd tests for in/out/seen and subscribe/notify/unsubscribe features.\n\nChange-Id: I15313f93e728a692820ac79644d70928d87b3f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/bf89dfbd788c6a1ac69d22edd3069284972fd34e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/bf89dfbd788c6a1ac69d22edd3069284972fd34e"}]},"branch":"refs/heads/master"},"658bf2d1bce477535b57384ccc8ddd3800db056b":{"kind":"REWORK","_number":2,"created":"2020-11-26 16:42:38.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/63/764363/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/63/764363/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/2"}}},"commit":{"parents":[{"commit":"8cc5cdd3a3ebc5559d94f567e90a2ceb1a3d0e54","subject":"Add tests for PTGbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/8cc5cdd3a3ebc5559d94f567e90a2ceb1a3d0e54"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-26 15:50:58.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-26 16:42:07.000000000","tz":60},"subject":"Add tests for user commands","message":"Add tests for user commands\n\nAdd tests for in/out/seen and subscribe/notify/unsubscribe features.\n\nChange-Id: I15313f93e728a692820ac79644d70928d87b3f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/658bf2d1bce477535b57384ccc8ddd3800db056b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/658bf2d1bce477535b57384ccc8ddd3800db056b"}]},"branch":"refs/heads/master"},"79b830ca3b4781b1e9bb05f1e5497a82507ee531":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-11-27 08:50:51.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/63/764363/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/63/764363/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/3"}}},"commit":{"parents":[{"commit":"5f371d18d401444a3c9ffd2283b6a8a268343b07","subject":"Add tests for PTGbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/5f371d18d401444a3c9ffd2283b6a8a268343b07"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-26 15:50:58.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-27 08:49:37.000000000","tz":60},"subject":"Add tests for user commands","message":"Add tests for user commands\n\nAdd tests for in/out/seen and subscribe/notify/unsubscribe features.\n\nChange-Id: I15313f93e728a692820ac79644d70928d87b3f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/79b830ca3b4781b1e9bb05f1e5497a82507ee531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/79b830ca3b4781b1e9bb05f1e5497a82507ee531"}]},"branch":"refs/heads/master"},"f368b40e98513362d305b592d67f358e5a452ef6":{"kind":"REWORK","_number":4,"created":"2020-12-08 13:36:36.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/63/764363/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/63/764363/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/63/764363/4"}}},"commit":{"parents":[{"commit":"5f371d18d401444a3c9ffd2283b6a8a268343b07","subject":"Add tests for PTGbot","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/5f371d18d401444a3c9ffd2283b6a8a268343b07"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-26 15:50:58.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-12-08 13:35:37.000000000","tz":60},"subject":"Add tests for user commands","message":"Add tests for user commands\n\nAdd tests for in/out/seen and subscribe/notify/unsubscribe features.\n\nChange-Id: I15313f93e728a692820ac79644d70928d87b3f2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/f368b40e98513362d305b592d67f358e5a452ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/f368b40e98513362d305b592d67f358e5a452ef6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
