)]}'
{"id":"openstack%2Fptgbot~764373","triplet_id":"openstack%2Fptgbot~master~I26e3683bbf3b7daab68407cd9928ee3d7f5c50dc","project":"openstack/ptgbot","branch":"master","topic":"refactor-cmds","hashtags":[],"change_id":"I26e3683bbf3b7daab68407cd9928ee3d7f5c50dc","subject":"Refactor processing of track and admin commands","status":"MERGED","created":"2020-11-26 16:58:26.000000000","updated":"2020-12-16 21:26:05.000000000","submitted":"2020-12-16 21:26:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"764373-refactor-cmds","meta_rev_id":"68d97b68bf5d6a1b6c84ebd237c516d7b5f38a19","_number":764373,"virtual_id_number":764373,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-16 21:26:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-16 21:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-12-16 21:19:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-26 17:10:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-16 21:19:36.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"079f66fdf75eb35e4fb8267677574619bff7a723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-11-26 16:58:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff84f21d08ea8f2a9b15ca338f1acbb657d7db34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 17:10:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae47036c34354f73a6f8683eb96eeee0 : SUCCESS in 3m 24s\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/9c4edfe559f640419a62f0f2c35b8fa4 : SUCCESS in 3m 11s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/fbb2907b90024aa9818647fe92a17350 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ca6959443d5bf2ad4b901966193f09fc721e8b2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-11-27 08:50:51.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"699c3791eebc13bae15fb97eed2d499c8c8a39e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 09:05:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35d0bdc986f54b1ea12714705afe7e99 : SUCCESS in 2m 49s\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/137fcf09c16344e1995642a9781389ef : SUCCESS in 2m 38s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/fd771db35532497e85a93f5a2634429a : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7896f30ca5800bba16eaf46033afbb0f998e1803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-12-08 13:36:36.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a8bf226712e9bce95f0f04b27937d19cbc6c4be5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-08 13:49:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b21ab43fe1443e8b29694f046d0802f : SUCCESS in 4m 02s\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/e72a86de28de43388cad848663cee965 : SUCCESS in 2m 39s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/92012028b84242e8907fcdfaaa33780b : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":3},{"id":"fac560a11d30739ecc62835184830362bba4fd59","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-12-16 21:19:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4106c98d40c81c7f225e1ee0ffd69fb2785ac0b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 21:19:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"46bfa3d4f528a1672375dd4d5ea38651ca5fccd4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 21:26:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/fd03044d7e48480f9498eb595b6cf356 : SUCCESS in 3m 47s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/79ab28dbc5924d1684cf15bd99d7facc : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":3},{"id":"68d97b68bf5d6a1b6c84ebd237c516d7b5f38a19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 21:26:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7beb7de516142a0cd2a8731232f1bc76e52e67f5","revisions":{"d68a0150f7ef6c02f7f9696aaeaeae10256d21bf":{"kind":"REWORK","_number":1,"created":"2020-11-26 16:58:26.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/73/764373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/73/764373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/1"}}},"commit":{"parents":[{"commit":"1d63a46f0b97f0c248b13870a883b9a9de2d27ea","subject":"Refactor processing of user commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/1d63a46f0b97f0c248b13870a883b9a9de2d27ea"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-26 16:56:18.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-26 16:56:18.000000000","tz":60},"subject":"Refactor processing of track and admin commands","message":"Refactor processing of track and admin commands\n\nSeparate the processing of track commands and admin commands\ninto separate files, and normalize handling of messages.\n\nAdjust test cases so that they match the new normalized messages.\n\nChange-Id: I26e3683bbf3b7daab68407cd9928ee3d7f5c50dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/d68a0150f7ef6c02f7f9696aaeaeae10256d21bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/d68a0150f7ef6c02f7f9696aaeaeae10256d21bf"}]},"branch":"refs/heads/master"},"e6c0bc3dfdcb8f400a50e1131cb8b87b879fe787":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-11-27 08:50:51.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/73/764373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/73/764373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/2"}}},"commit":{"parents":[{"commit":"4058571a2856707978ed134360eccf43e30b2b92","subject":"Refactor processing of user commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/4058571a2856707978ed134360eccf43e30b2b92"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-26 16:56:18.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-27 08:49:37.000000000","tz":60},"subject":"Refactor processing of track and admin commands","message":"Refactor processing of track and admin commands\n\nSeparate the processing of track commands and admin commands\ninto separate files, and normalize handling of messages.\n\nAdjust test cases so that they match the new normalized messages.\n\nChange-Id: I26e3683bbf3b7daab68407cd9928ee3d7f5c50dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/e6c0bc3dfdcb8f400a50e1131cb8b87b879fe787"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/e6c0bc3dfdcb8f400a50e1131cb8b87b879fe787"}]},"branch":"refs/heads/master"},"7beb7de516142a0cd2a8731232f1bc76e52e67f5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-12-08 13:36:36.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/73/764373/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/73/764373/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/73/764373/3"}}},"commit":{"parents":[{"commit":"2f8c18d9c10c5acebc239140ac79c630114fc192","subject":"Refactor processing of user commands","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/2f8c18d9c10c5acebc239140ac79c630114fc192"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-11-26 16:56:18.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-12-08 13:35:48.000000000","tz":60},"subject":"Refactor processing of track and admin commands","message":"Refactor processing of track and admin commands\n\nSeparate the processing of track commands and admin commands\ninto separate files, and normalize handling of messages.\n\nAdjust test cases so that they match the new normalized messages.\n\nChange-Id: I26e3683bbf3b7daab68407cd9928ee3d7f5c50dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/7beb7de516142a0cd2a8731232f1bc76e52e67f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/7beb7de516142a0cd2a8731232f1bc76e52e67f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
