)]}'
{"id":"openstack%2Fptgbot~798025","triplet_id":"openstack%2Fptgbot~master~If146ddc6176485cab038e68fae5c7cdbd7202417","project":"openstack/ptgbot","branch":"master","topic":"containerize","hashtags":[],"change_id":"If146ddc6176485cab038e68fae5c7cdbd7202417","subject":"Add Container Image Build","status":"MERGED","created":"2021-06-24 20:48:18.000000000","updated":"2021-07-06 23:27:11.000000000","submitted":"2021-07-06 23:25:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"798025-containerize","meta_rev_id":"48c9a519496e34cee72152dbce67e4ba6befd12c","_number":798025,"virtual_id_number":798025,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-06 23:25:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-06 23:15:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"all":[{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-06 23:15:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-24 20:48:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-24 20:53:43.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2021-06-24 20:53:43.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2021-06-24 20:53:43.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-06-24 20:53:43.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"state":"REVIEWER"}],"messages":[{"id":"6bbf9cd6d13aaf5c271e51d532d3e951e4150e92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-06-24 20:48:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"150a6de1f454d732073626b28647328b4b4076ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 20:48:38.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ptgbot on branch master.  The error was:\n\n  while scanning a plain scalar\n  in \"openstack/ptgbot/.zuul.yaml@master\", line 43, column 11\nfound a tab character that violates indentation\n  in \"openstack/ptgbot/.zuul.yaml@master\", line 44, column 1","accounts_in_message":[],"_revision_number":1},{"id":"29a2c7cc6b88ec5cdc745e56b0eb680660a4bb49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-06-24 20:50:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dacc88d304d5aec9682cee8ff251fb02cb1db04b","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-06-24 20:52:26.000000000","message":"Patch Set 2:\n\nI need a fancy infra friend to generate a secret for me :)","accounts_in_message":[],"_revision_number":2},{"id":"321a499170a5194616cd68ad8760cda319cf210c","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-06-24 20:52:34.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"fb990b951083ca67fb74cee48fcbc03f77d63b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 20:58:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/73f20092b333409c9df8b80c41bbcf86 : SUCCESS in 5m 24s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/48f2395bd73b4706abbd63afab4998fc : SUCCESS in 5m 27s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/c4c6b8bf142346dd8a5998f9b628247c : FAILURE in 6m 12s","accounts_in_message":[],"_revision_number":2},{"id":"4eea84b2ffba7edab62b3456dbf332b75e670c89","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-24 21:49:49.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9bd8bd0fde0c673dfa9845c193b8daeec7bb0fa3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-06-24 22:14:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d61fa1fb9ba13610c8bdc97ed81f0138bb7d6df","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-06-28 21:41:09.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"637a341ec8ddc05d1f769441311353371254d9c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-06-28 21:49:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"051d59f4eca162a9b3f71d5c018ed9993e7ab1cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 21:57:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/7b99bef5990a44d295828ce34a1dccd1 : SUCCESS in 5m 34s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/a33147cb18704b3d9324a1314d0ecf9b : SUCCESS in 5m 42s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/5307181be39840cbb6e329642a40a34c : FAILURE in 6m 57s","accounts_in_message":[],"_revision_number":3},{"id":"856bc4c35068751cc9667b1cd412a2798a029a64","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-29 19:48:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ee420a833f21944eb47a55067ea1ef2231874d5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-06-29 19:54:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"eb63a5e29fd6c295eb70baf77a07049dc9d8551f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 20:06:00.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/593c98f93369454e9c6cc02ca925cd59 : SUCCESS in 4m 51s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/fc075ba3ead24ba1927097eec81094cd : SUCCESS in 4m 58s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/37fd8fe8ccb84dd2b7c11bf143ee3272 : FAILURE in 5m 46s","accounts_in_message":[],"_revision_number":4},{"id":"7f4f7f46ae4666a1a243b54d1d965dc4e092fe9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-06-29 20:21:45.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"0110b48d2e9c078d31ea82eb5f930d9e0c348093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 20:33:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/26efaf7c641c4c77af9bf9c9569aaef4 : SUCCESS in 4m 47s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/6ea14a4cc0d24407a6a483a28ce1e2d1 : SUCCESS in 5m 07s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/0fbaec34cbc844139b154d2ded229837 : FAILURE in 5m 55s","accounts_in_message":[],"_revision_number":5},{"id":"e25db8b2926a960dd567f8476274b63fac621dc7","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-29 20:37:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3c2a3090bbd44365cf269c98094b1c4b386f5af3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-06-29 20:42:44.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"a9d935852eff0680aae42c36506accbf4f8b2c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 20:52:31.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/03f5eba2c9dd497cbf1d4e898faf3fd1 : SUCCESS in 4m 53s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/85a7affff8c642e2a53a831a5a50c477 : SUCCESS in 5m 08s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/8acde150e7664764b9537325f861836a : FAILURE in 6m 08s","accounts_in_message":[],"_revision_number":6},{"id":"bba64fe80e1bebe7ca557b4ccc6eb2024843a8f4","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-06-29 20:54:42.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1908ed2ef39043a9f162f000b420fb21ee466b6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-06-30 20:35:38.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"9b3fd989eaa920a0317f51c5ab7095d026b15f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 20:46:33.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/da98083da6d5456e9de07cca7268b161 : SUCCESS in 4m 23s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/0165759e0912407b9b41886ca05aff32 : SUCCESS in 4m 43s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/73613fba118a436aa2df5bd1bbbee431 : SUCCESS in 6m 13s","accounts_in_message":[],"_revision_number":7},{"id":"689d3fbf439affa1fa596bb9f0d03b9c83e78f15","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-07-01 00:28:31.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"153fc4533ea07e402abb1ed5858dd3bc9eb3d12e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 00:29:09.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"0bd7a674cac4f089f57ad4a72f50fbdd4894ecd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 00:40:14.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/1be95b5fb41b4d7ca3618bb2bb90364b : SUCCESS in 4m 44s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/f48ffb6d6c64421499617525b6f0a7ec : SUCCESS in 3m 48s\n- ptgbot-upload-opendev-image https://zuul.opendev.org/t/openstack/build/3ca4b328c81741419d7a5e835c7c8d1f : POST_FAILURE in 6m 47s","accounts_in_message":[],"_revision_number":7},{"id":"73ebeb2e0f6d0a151a4704c8d482460db7a3232f","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-07-01 05:02:19.000000000","message":"Patch Set 7: Code-Review-1 -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ceaf9ecd5ac74a9ecf137e6a8e4cd191c75943bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2021-07-06 22:14:17.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"06a35b733f8c5f40faf93924548ee5cd3d2e0e51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 22:24:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/4e2d22248f194e7ba1346fbf0bd96cc0 : SUCCESS in 3m 31s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/a2f19cfa399c459fa714cda30821f84c : SUCCESS in 3m 26s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/9f3452c9f9c1491db9e23392b4d095f5 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":8},{"id":"828b06796fb59cb05114acae43dbf93c7722e5e6","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-07-06 23:15:01.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"001a3895a14ada520c47c349c8f880e4d4181717","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 23:15:22.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"bd32bca37452064e43bc7dcd5dd0aec5cd0ecfaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 23:25:44.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/38231f8b2df74777aab3f7d0b96d7782 : SUCCESS in 3m 18s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/f74fbec646ff4333a05b9f4ca56077fa : SUCCESS in 3m 34s\n- ptgbot-upload-opendev-image https://zuul.opendev.org/t/openstack/build/79139160ce5145a38015ac8b73e4d4f7 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":8},{"id":"6e0185384700819c81773fc52b84d2c4ab1f4673","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 23:25:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"48c9a519496e34cee72152dbce67e4ba6befd12c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 23:27:11.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- ptgbot-promote-opendev-image https://zuul.opendev.org/t/openstack/build/a0e2a5959cd348dda155cba4701ce18b : SUCCESS in 43s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"a3709343151ca56d29aaa50c84fb9a6789730fdd","revisions":{"87b38a74dc4f1f58063f306240c1aa4986ac84a0":{"kind":"REWORK","_number":1,"created":"2021-06-24 20:48:18.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/25/798025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/25/798025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/1"}}},"commit":{"parents":[{"commit":"6445b85ccdb920df56f8a081e65dfdd9e3903793","subject":"Add missing clear clean alias","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/6445b85ccdb920df56f8a081e65dfdd9e3903793"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-24 20:48:01.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-24 20:48:01.000000000","tz":-420},"subject":"Add Container Image Build","message":"Add Container Image Build\n\nChange-Id: If146ddc6176485cab038e68fae5c7cdbd7202417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/87b38a74dc4f1f58063f306240c1aa4986ac84a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/87b38a74dc4f1f58063f306240c1aa4986ac84a0"}]},"branch":"refs/heads/master"},"d7c803897951ea195ab323cda51246481c1f856b":{"kind":"REWORK","_number":2,"created":"2021-06-24 20:50:08.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/25/798025/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/25/798025/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/2"}}},"commit":{"parents":[{"commit":"6445b85ccdb920df56f8a081e65dfdd9e3903793","subject":"Add missing clear clean alias","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/6445b85ccdb920df56f8a081e65dfdd9e3903793"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-24 20:48:01.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-24 20:50:01.000000000","tz":-420},"subject":"Add Container Image Build","message":"Add Container Image Build\n\nChange-Id: If146ddc6176485cab038e68fae5c7cdbd7202417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/d7c803897951ea195ab323cda51246481c1f856b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/d7c803897951ea195ab323cda51246481c1f856b"}]},"branch":"refs/heads/master"},"627d7e53951d1d2bf85c0e9a81be34097ad891d7":{"kind":"REWORK","_number":3,"created":"2021-06-28 21:49:52.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/25/798025/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/25/798025/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/3"}}},"commit":{"parents":[{"commit":"6445b85ccdb920df56f8a081e65dfdd9e3903793","subject":"Add missing clear clean alias","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/6445b85ccdb920df56f8a081e65dfdd9e3903793"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-24 20:48:01.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-28 21:49:46.000000000","tz":-420},"subject":"Add Container Image Build","message":"Add Container Image Build\n\nChange-Id: If146ddc6176485cab038e68fae5c7cdbd7202417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/627d7e53951d1d2bf85c0e9a81be34097ad891d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/627d7e53951d1d2bf85c0e9a81be34097ad891d7"}]},"branch":"refs/heads/master"},"622ca5251f5314ce4a40748d7f33ac2a6c067355":{"kind":"REWORK","_number":4,"created":"2021-06-29 19:54:11.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/25/798025/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/25/798025/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/4"}}},"commit":{"parents":[{"commit":"6445b85ccdb920df56f8a081e65dfdd9e3903793","subject":"Add missing clear clean alias","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/6445b85ccdb920df56f8a081e65dfdd9e3903793"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-24 20:48:01.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-29 19:53:32.000000000","tz":-420},"subject":"Add Container Image Build","message":"Add Container Image Build\n\nChange-Id: If146ddc6176485cab038e68fae5c7cdbd7202417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/622ca5251f5314ce4a40748d7f33ac2a6c067355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/622ca5251f5314ce4a40748d7f33ac2a6c067355"}]},"branch":"refs/heads/master"},"fbc802b74c22a00ded02efba76770784a8fad21c":{"kind":"REWORK","_number":5,"created":"2021-06-29 20:21:45.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/25/798025/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/25/798025/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/5"}}},"commit":{"parents":[{"commit":"6445b85ccdb920df56f8a081e65dfdd9e3903793","subject":"Add missing clear clean alias","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/6445b85ccdb920df56f8a081e65dfdd9e3903793"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-24 20:48:01.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-29 20:21:26.000000000","tz":0},"subject":"Add Container Image Build","message":"Add Container Image Build\n\nChange-Id: If146ddc6176485cab038e68fae5c7cdbd7202417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/fbc802b74c22a00ded02efba76770784a8fad21c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/fbc802b74c22a00ded02efba76770784a8fad21c"}]},"branch":"refs/heads/master"},"d2af9da414baee2f70aac768d6b4e231281c28a9":{"kind":"REWORK","_number":6,"created":"2021-06-29 20:42:44.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/25/798025/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/25/798025/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/6"}}},"commit":{"parents":[{"commit":"6445b85ccdb920df56f8a081e65dfdd9e3903793","subject":"Add missing clear clean alias","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/6445b85ccdb920df56f8a081e65dfdd9e3903793"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-24 20:48:01.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-29 20:42:28.000000000","tz":0},"subject":"Add Container Image Build","message":"Add Container Image Build\n\nChange-Id: If146ddc6176485cab038e68fae5c7cdbd7202417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/d2af9da414baee2f70aac768d6b4e231281c28a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/d2af9da414baee2f70aac768d6b4e231281c28a9"}]},"branch":"refs/heads/master"},"d4d03aa7eac4e30893f148217dd388efb3625af8":{"kind":"REWORK","_number":7,"created":"2021-06-30 20:35:38.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/25/798025/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/25/798025/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/7"}}},"commit":{"parents":[{"commit":"6445b85ccdb920df56f8a081e65dfdd9e3903793","subject":"Add missing clear clean alias","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/6445b85ccdb920df56f8a081e65dfdd9e3903793"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-24 20:48:01.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-30 20:35:31.000000000","tz":0},"subject":"Add Container Image Build","message":"Add Container Image Build\n\nChange-Id: If146ddc6176485cab038e68fae5c7cdbd7202417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/d4d03aa7eac4e30893f148217dd388efb3625af8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/d4d03aa7eac4e30893f148217dd388efb3625af8"}]},"branch":"refs/heads/master"},"a3709343151ca56d29aaa50c84fb9a6789730fdd":{"kind":"REWORK","_number":8,"created":"2021-07-06 22:14:17.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/25/798025/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/25/798025/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/25/798025/8"}}},"commit":{"parents":[{"commit":"6445b85ccdb920df56f8a081e65dfdd9e3903793","subject":"Add missing clear clean alias","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/6445b85ccdb920df56f8a081e65dfdd9e3903793"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-06-24 20:48:01.000000000","tz":-420},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2021-07-06 22:14:13.000000000","tz":0},"subject":"Add Container Image Build","message":"Add Container Image Build\n\nChange-Id: If146ddc6176485cab038e68fae5c7cdbd7202417\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/a3709343151ca56d29aaa50c84fb9a6789730fdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/a3709343151ca56d29aaa50c84fb9a6789730fdd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
