)]}'
{"id":"openstack%2Fptgbot~812417","triplet_id":"openstack%2Fptgbot~master~I86f3f5002f88005760e00c5714b0891c1892a6f2","project":"openstack/ptgbot","branch":"master","topic":"serve-static","hashtags":[],"change_id":"I86f3f5002f88005760e00c5714b0891c1892a6f2","subject":"ptgbot-web: add simple service to serve static pages","status":"MERGED","created":"2021-10-05 01:51:35.000000000","updated":"2021-10-05 14:24:21.000000000","submitted":"2021-10-05 14:18:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812417-serve-static","meta_rev_id":"c90fedec28a7f026c176dea4a1f0dccf7e0c6059","_number":812417,"virtual_id_number":812417,"owner":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-05 14:18:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2021-10-05 11:54:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-05 12:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-05 13:55:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-05 01:58:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-05 02:05:49.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2021-10-05 11:54:24.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"bf6410d0129463bd2de6c3d55e4b3ff0e2469196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-10-05 01:51:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eaa3c7cd2890c89c226f8484cb437c3721f77ba0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 01:58:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/0ac519591519406e9cb4333c393441c0 : SUCCESS in 3m 24s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/2094ee3a1c9345e38fdf5387e59ded37 : SUCCESS in 4m 18s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/6fd93e2259a447ddaa875662390536f1 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"b1590ab392cb3344acb60111b4e9fc2907a809c3","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-10-05 02:05:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"668c9d9b04b1100cc847b604c7e4fca8bd1a0791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-10-05 02:53:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d78e0867b1487e1bb8502035630874ad9153b75d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 03:04:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/4e313e82511a4554a34746a1c004b8cf : SUCCESS in 3m 07s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/2ef115962e844f58b0bda7b5f3b7f417 : SUCCESS in 3m 29s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/68a73e11694f48c8baf0edbac003374f : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"78fc78ccf50476ed24052622ad0d40cdc1032c69","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-10-05 03:14:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"51dbeee98831708a3b0e680d211caaeabfe87194","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"date":"2021-10-05 03:39:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bb7b6da03fe66e9675b5db989682d13ad3a9300a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 03:47:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/a123cde92910490fb35874dc4f6f9787 : SUCCESS in 3m 31s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/2005c22a176e432a8cc6627ea29e717c : SUCCESS in 3m 44s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/27450e0f4f8944bb8f3baf878848c48c : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":3},{"id":"0ba63a6bab2488dc46a6eccbc14afc351ecc8239","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2021-10-05 11:54:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"6546badc029646d1a9c3600f9c982e191fc4cbef","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-10-05 12:39:32.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ebf89fc15bf40a6f13d35825d80ef0e4e074375b","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2021-10-05 13:55:55.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9f392a7d385b1b0ba06f2d4b17fe27758e9ba9de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 14:06:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3c61afe8384a3c8c767b40ed6f1d18a84a849f94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 14:18:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/d440f026ae3a41b1a27e2aecfa455d12 : SUCCESS in 4m 21s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/6ed61b264ab6421b9c90fa880fd53510 : SUCCESS in 4m 33s\n- ptgbot-upload-opendev-image https://zuul.opendev.org/t/openstack/build/b8a8a2af38e2472badddd277cb140437 : SUCCESS in 7m 05s","accounts_in_message":[],"_revision_number":3},{"id":"61be6e88bd466a242ae9d6242e1ae78cc8755fa7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 14:18:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c90fedec28a7f026c176dea4a1f0dccf7e0c6059","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 14:24:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- ptgbot-promote-opendev-image https://zuul.opendev.org/t/openstack/build/36256969fafc46f99af9778e03f6dc86 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0e2f3d0a9d935c9b6a2bfeb1480d903ed4aab513","revisions":{"0122c30cdd058fb95fef1bc6232176fffa292822":{"kind":"REWORK","_number":1,"created":"2021-10-05 01:51:35.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/17/812417/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/17/812417/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/1"}}},"commit":{"parents":[{"commit":"82c0b9ab9dd3325767dfdf1044576931111f7081","subject":"Merge \"Adapt PTGBot to OFTC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/82c0b9ab9dd3325767dfdf1044576931111f7081"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-05 01:50:41.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-05 01:51:32.000000000","tz":660},"subject":"ptgbot-web: add simple service to serve static pages","message":"ptgbot-web: add simple service to serve static pages\n\nThis serves the static pages from the container directly\n\nChange-Id: I86f3f5002f88005760e00c5714b0891c1892a6f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/0122c30cdd058fb95fef1bc6232176fffa292822"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/0122c30cdd058fb95fef1bc6232176fffa292822"}]},"branch":"refs/heads/master"},"ddf4adfdf887482527ccb504373306524559351f":{"kind":"REWORK","_number":2,"created":"2021-10-05 02:53:11.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/17/812417/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/17/812417/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/2"}}},"commit":{"parents":[{"commit":"82c0b9ab9dd3325767dfdf1044576931111f7081","subject":"Merge \"Adapt PTGBot to OFTC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/82c0b9ab9dd3325767dfdf1044576931111f7081"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-05 01:50:41.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-05 02:52:59.000000000","tz":660},"subject":"ptgbot-web: add simple service to serve static pages","message":"ptgbot-web: add simple service to serve static pages\n\nThis serves the static pages from the container directly\n\nChange-Id: I86f3f5002f88005760e00c5714b0891c1892a6f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/ddf4adfdf887482527ccb504373306524559351f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/ddf4adfdf887482527ccb504373306524559351f"}]},"branch":"refs/heads/master"},"0e2f3d0a9d935c9b6a2bfeb1480d903ed4aab513":{"kind":"REWORK","_number":3,"created":"2021-10-05 03:39:08.000000000","uploader":{"_account_id":7118,"name":"Ian Wienand","email":"iwienand@redhat.com","username":"iwienand"},"ref":"refs/changes/17/812417/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/17/812417/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/17/812417/3"}}},"commit":{"parents":[{"commit":"82c0b9ab9dd3325767dfdf1044576931111f7081","subject":"Merge \"Adapt PTGBot to OFTC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/82c0b9ab9dd3325767dfdf1044576931111f7081"}]}],"author":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-05 01:50:41.000000000","tz":660},"committer":{"name":"Ian Wienand","email":"iwienand@redhat.com","date":"2021-10-05 03:38:56.000000000","tz":660},"subject":"ptgbot-web: add simple service to serve static pages","message":"ptgbot-web: add simple service to serve static pages\n\nThis serves the static pages from the container directly\n\nChange-Id: I86f3f5002f88005760e00c5714b0891c1892a6f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/0e2f3d0a9d935c9b6a2bfeb1480d903ed4aab513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/0e2f3d0a9d935c9b6a2bfeb1480d903ed4aab513"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
