)]}'
{"id":"openstack%2Fptgbot~841575","triplet_id":"openstack%2Fptgbot~master~Iac2201b9220db2c9b4c07a72cae1437b3f97de3a","project":"openstack/ptgbot","branch":"master","topic":"room_details","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2022-05-13 09:31:30.000000000","reason":"Change was submitted"},"16708":{"account":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"last_update":"2022-05-13 09:31:30.000000000","reason":"Change was submitted"},"308":{"account":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"last_update":"2022-05-13 09:24:29.000000000","reason":"Thierry Carrez replied on the change"},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-05-12 12:21:56.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Iac2201b9220db2c9b4c07a72cae1437b3f97de3a","subject":"Update Bot to Show Room Descriptions","status":"MERGED","created":"2022-05-12 11:15:23.000000000","updated":"2022-05-13 09:32:34.000000000","submitted":"2022-05-13 09:31:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841575-room_details","meta_rev_id":"0c6a1c2773edbd762994adf3ba7a3e1ba388ecb5","_number":841575,"virtual_id_number":841575,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-13 09:31:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2022-05-13 09:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-05-12 12:21:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2022-05-13 09:24:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"}],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-12 11:19:05.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2022-05-12 11:19:05.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2022-05-12 11:19:05.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2022-05-12 11:25:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ff775dd1b45f21449d2713b69f54d78158d2da5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2022-05-12 11:15:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42958a285d8e6ca4c878251aea1f4ff6a5b17ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 11:25:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/aff932eafbe54e40a657b48fc4c1e972 : SUCCESS in 2m 56s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/2e9d33dfe1b54fbbaaf826262d2b81d1 : SUCCESS in 3m 04s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/c66e9f8ce2ac4245828da24ca5b8045d : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"54e9ae1881827ecddf56c8d259bb29a4c4fe38b0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-05-12 11:32:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbcff3702065548053333404bce223287d0055c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2022-05-12 11:44:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3f3dd4e3237e5c01ef08b70ca0a5ca28f9873615","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2022-05-12 11:45:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c8c3c7f3fbe3822d52f677deed62ccd1d0d632e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 11:55:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/6e177c03da7d4058b85970a3f0ed343a : SUCCESS in 2m 42s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/fe802eee6112444fae42cd9eaa3cd4bc : SUCCESS in 2m 58s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/5c12f730c74c4dd2bc00356e67e222ba : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":2},{"id":"10773b026e633847206efbfd9da14b0af201f3a8","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-05-12 12:21:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"19de869fbb3c470719570ee52a47a7e9b26bf664","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-05-13 09:24:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f912bcf9aaee434f6b1328920f26b77241368318","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 09:24:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a7e324373d5b362352c6dcb250c3d3f6f6f1b689","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 09:31:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/b70cacdcc7b6461f94a1ff8842e8b3ac : SUCCESS in 3m 41s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/7966cd68171242829efb2ae8e953276e : SUCCESS in 2m 43s\n- ptgbot-upload-opendev-image https://zuul.opendev.org/t/openstack/build/006127b8bb7443feb84ac99f961ff362 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"e7f693f237f615e1faec2b48c007c60bcf2fadda","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 09:31:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0c6a1c2773edbd762994adf3ba7a3e1ba388ecb5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 09:32:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- ptgbot-promote-opendev-image https://zuul.opendev.org/t/openstack/build/d982084b9013422cad4d65dbc432210c : SUCCESS in 44s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8ddd44bb1ccaf5fa468e0fa8d1dd4d129d8f46a7","revisions":{"23ba456670e2812c29c89e8a055aa06540e45f08":{"kind":"REWORK","_number":1,"created":"2022-05-12 11:15:23.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/75/841575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/75/841575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/75/841575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/75/841575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/75/841575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/75/841575/1"}}},"commit":{"parents":[{"commit":"b072d88c0a3b049c39276af64181104efbff748b","subject":"Update ptgbot\u0027s docker image to bullseye","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/b072d88c0a3b049c39276af64181104efbff748b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2022-05-12 11:15:01.000000000","tz":-300},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2022-05-12 11:15:01.000000000","tz":-300},"subject":"Update Bot to Show Room Descriptions","message":"Update Bot to Show Room Descriptions\n\nChange-Id: Iac2201b9220db2c9b4c07a72cae1437b3f97de3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/23ba456670e2812c29c89e8a055aa06540e45f08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/23ba456670e2812c29c89e8a055aa06540e45f08"}]},"branch":"refs/heads/master"},"8ddd44bb1ccaf5fa468e0fa8d1dd4d129d8f46a7":{"kind":"REWORK","_number":2,"created":"2022-05-12 11:44:39.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/75/841575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/75/841575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/75/841575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/75/841575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/75/841575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/75/841575/2"}}},"commit":{"parents":[{"commit":"b072d88c0a3b049c39276af64181104efbff748b","subject":"Update ptgbot\u0027s docker image to bullseye","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/b072d88c0a3b049c39276af64181104efbff748b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2022-05-12 11:15:01.000000000","tz":-300},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2022-05-12 11:44:05.000000000","tz":-300},"subject":"Update Bot to Show Room Descriptions","message":"Update Bot to Show Room Descriptions\n\nAlso adds examples of text for the description field\nto base.json\n\nChange-Id: Iac2201b9220db2c9b4c07a72cae1437b3f97de3a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/8ddd44bb1ccaf5fa468e0fa8d1dd4d129d8f46a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/8ddd44bb1ccaf5fa468e0fa8d1dd4d129d8f46a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
