)]}'
{"id":"openstack%2Fptgbot~858100","triplet_id":"openstack%2Fptgbot~master~I69d5b79d8d894f94919724c0f94225dbf959b20e","project":"openstack/ptgbot","branch":"master","topic":"command-docs","attention_set":{},"removed_from_attention_set":{"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2022-09-20 12:16:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69d5b79d8d894f94919724c0f94225dbf959b20e","subject":"Clarify required parameter names in readme","status":"MERGED","created":"2022-09-16 11:47:37.000000000","updated":"2022-09-20 12:17:47.000000000","submitted":"2022-09-20 12:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"858100-command-docs","meta_rev_id":"809223cd6b8f0479d15f117466c04d8aa9684587","_number":858100,"virtual_id_number":858100,"owner":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-20 12:16:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-09-16 12:02:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-09-20 12:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-09-20 12:03:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-16 11:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-16 12:02:50.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-09-20 12:03:44.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"cfa8ff8cf0ff69acbede567dacdc1c38167dc2f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2022-09-16 11:47:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1ba5a8731b3688211fff21449838f9e3cafdb61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-16 11:59:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b078e029352146f890ad7ddfc85ce5cc\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/9d17ef0b6cc24862be695eee139ad573 : SUCCESS in 3m 23s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/60f995011151495a831196bf90a8b197 : SUCCESS in 3m 34s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/ce5eedb2633a4c72bd14c5db3ea10d3b : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0c563cdf22ff3eea2017e65d1679d7f2fcfe1f43","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-09-16 12:02:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c223a927058ecf58e4ce970bcbb461467b49d9e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2022-09-20 12:03:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"addf1f7656e6f25524efa58ead8c5ecad4d5f13e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 12:04:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0b11a29417f9465db6b30533fcd26d0e22c58c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 12:16:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/006ad08c247d4a04a1084a84f488e1c7\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/d1a498a6dd2647c7ae715f8269410c4e : SUCCESS in 2m 55s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/3f8513b2054749999fb90daffaf209c2 : SUCCESS in 3m 26s\n- ptgbot-upload-opendev-image https://zuul.opendev.org/t/openstack/build/1d2362c74abf4e4890603b643db0cddb : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"513104c5a96904b4716503fbe70ce77c7d8b61e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 12:16:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"809223cd6b8f0479d15f117466c04d8aa9684587","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-20 12:17:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55d598ad3b0c40ce9a2de1cc348943f0\n\n- ptgbot-promote-opendev-image https://zuul.opendev.org/t/openstack/build/a7f8fcc42417461ea2b8877b58f9486f : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b6b8293a81b229584106502d27f8df1608237cd","revisions":{"7b6b8293a81b229584106502d27f8df1608237cd":{"kind":"REWORK","_number":1,"created":"2022-09-16 11:47:37.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/00/858100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/00/858100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/00/858100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/00/858100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/00/858100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/00/858100/1"}}},"commit":{"parents":[{"commit":"8ddd44bb1ccaf5fa468e0fa8d1dd4d129d8f46a7","subject":"Update Bot to Show Room Descriptions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/8ddd44bb1ccaf5fa468e0fa8d1dd4d129d8f46a7"}]}],"author":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-09-16 11:46:02.000000000","tz":0},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2022-09-16 11:46:02.000000000","tz":0},"subject":"Clarify required parameter names in readme","message":"Clarify required parameter names in readme\n\nMultiple people have been mistaking the ~add and ~del syntax to\ninclude a keyword \"TRACK\" which results in creating an actual track\nby that name. Add chevrons around required parameters in the syntax\ndocumentation in order to make it clear those are parameter names\nand not literal keywords.\n\nChange-Id: I69d5b79d8d894f94919724c0f94225dbf959b20e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/7b6b8293a81b229584106502d27f8df1608237cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/7b6b8293a81b229584106502d27f8df1608237cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}}]}],"submit_requirements":[]}
