)]}'
{"id":"openstack%2Fptgbot~885499","triplet_id":"openstack%2Fptgbot~master~Ib6946b974ebdce7ada55ed39d750e54a9e1e83d6","project":"openstack/ptgbot","branch":"master","topic":"feature/iCal","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-09-25 00:55:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib6946b974ebdce7ada55ed39d750e54a9e1e83d6","subject":"Add stestr to .gitignore","status":"MERGED","created":"2023-06-07 14:47:57.000000000","updated":"2023-09-25 00:56:12.000000000","submitted":"2023-09-25 00:55:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885499-feature/iCal","meta_rev_id":"ffc3988c1fb8a469994610e56103764ca0b987b2","_number":885499,"virtual_id_number":885499,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-25 00:55:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-09-25 00:21:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-25 00:49:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-25 00:49:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-07 15:01:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-08 09:17:29.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-09-25 00:49:45.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"859c3a2318cfb54c97d80aa125a40e43a7ac4eb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-06-07 14:47:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"792856f9bd049f64421fb5027792310914cd5e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 15:01:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8af290ae8b6f41298ef762c704562f05\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/1e59211a95224f14b7fa80bab9526f1e : SUCCESS in 2m 25s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/eb1a3c0d9e994e2a9dbfdebda97c8150 : SUCCESS in 2m 33s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/625617a01dae45cba751055d680f0b9b : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5b14427a9a57fafef32387d5bb65ec8b3a550804","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-06-08 09:17:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e0f483ff150f69904c20668b200dfe4718cae700","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-09-25 00:21:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3cdb814cad3af618bc61745d2c3b7274bdb7d308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 00:26:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b42a4ed56dee4e1584c5bc1a9f30a8ec\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/47f748c4662f4482a778e66e5f893f8c : SUCCESS in 2m 49s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/88376bd1d99f4c18a8289377222ec65a : SUCCESS in 2m 30s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/1b47cfef3aa043ac890bb440cd1a409e : SUCCESS in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ec66fed474d105b2f75087762d9d8ced56b2a578","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-25 00:49:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"17637d88fe339b66d795225eb7b474b585e2076e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 00:50:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e67c558f5281651533386a923085f296397bd804","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 00:55:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d455f46d12d14f0a8ce21805a5cc19a1\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/e50f524c5c0a479d921a6dd4a3d7d7df : SUCCESS in 2m 10s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/981941c1595f40848cab78f1daeb680d : SUCCESS in 2m 29s\n- ptgbot-upload-opendev-image https://zuul.opendev.org/t/openstack/build/264178252c6b43c690bb4d5fe29bb474 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"522093c595c352aa941293a2a29469a0cc65b6ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 00:55:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ffc3988c1fb8a469994610e56103764ca0b987b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 00:56:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fd05765f2434672bc42c873c79dd3d6\n\n- ptgbot-promote-opendev-image https://zuul.opendev.org/t/openstack/build/0967f3f75ecb404bb74be01e9b637e81 : SUCCESS in 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"73e7258e6df33121c78a6f1ebae14ce763a1ab4a","revisions":{"9585d97a80d0a9089145cac86af4fc2195f5e2c9":{"kind":"REWORK","_number":1,"created":"2023-06-07 14:47:57.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/99/885499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/99/885499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/99/885499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/99/885499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/99/885499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/99/885499/1"}}},"commit":{"parents":[{"commit":"70838b9d9328730ee6740838793d070cdaaf278c","subject":"Revert \"Migrate ptgbot to quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/70838b9d9328730ee6740838793d070cdaaf278c"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-06-07 14:19:13.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-06-07 14:47:36.000000000","tz":-300},"subject":"Add stestr to .gitignore","message":"Add stestr to .gitignore\n\nChange-Id: Ib6946b974ebdce7ada55ed39d750e54a9e1e83d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/9585d97a80d0a9089145cac86af4fc2195f5e2c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/9585d97a80d0a9089145cac86af4fc2195f5e2c9"}]},"branch":"refs/heads/master"},"73e7258e6df33121c78a6f1ebae14ce763a1ab4a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-25 00:21:22.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/99/885499/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/99/885499/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/99/885499/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/99/885499/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/99/885499/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/99/885499/2"}}},"commit":{"parents":[{"commit":"513944afea7aa61110a2d8b30bcf3fc7bdb85c2a","subject":"Update ptgbot to python3.11 and Bookworm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/513944afea7aa61110a2d8b30bcf3fc7bdb85c2a"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-06-07 14:19:13.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-09-25 00:21:12.000000000","tz":600},"subject":"Add stestr to .gitignore","message":"Add stestr to .gitignore\n\nChange-Id: Ib6946b974ebdce7ada55ed39d750e54a9e1e83d6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/73e7258e6df33121c78a6f1ebae14ce763a1ab4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/73e7258e6df33121c78a6f1ebae14ce763a1ab4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
