)]}'
{"id":"openstack%2Fptgbot~885502","triplet_id":"openstack%2Fptgbot~master~I2accbc0db25d2e2d0700aad27b4e31de7ba63628","project":"openstack/ptgbot","branch":"master","topic":"feature/iCal","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2023-10-03 12:19:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2accbc0db25d2e2d0700aad27b4e31de7ba63628","subject":"[scripts]  Add an example/helper script for adding realtime fields.","status":"MERGED","created":"2023-06-07 14:47:57.000000000","updated":"2023-10-03 12:20:48.000000000","submitted":"2023-10-03 12:19:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885502-feature/iCal","meta_rev_id":"798933cec261efe8454da28a21c4002f39b5d28d","_number":885502,"virtual_id_number":885502,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-03 12:19:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2023-09-25 02:01:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-03 12:11:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-03 12:11:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-07 14:59:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-08 09:19:43.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2023-10-03 12:11:35.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"118e1ad9a0c5b8a6284a8e99097042d0f8bc8405","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-06-07 14:47:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebb3729c5415a68e3c762b813d660d981408d7f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 14:59:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16bb7597189c48d99c34889d87c1f83f\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/bcc66225bd4d495b85c74a568256c0d9 : SUCCESS in 2m 41s\n- tox-py38 https://zuul.opendev.org/t/openstack/build/a38508f2fa4547d2ac209e3f8198f355 : SUCCESS in 2m 39s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/6322bfebe71745eab9fc0698503125d2 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"832cca6b9ea2da13a814f12467c5f949d8bd29b7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2023-06-08 09:19:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b958f51969e5bddba8f1108bb47307aa8a7ec533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-09-25 00:21:22.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"be53e33383c4e21a247bb8f82fcabda13ec8e3a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 00:33:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8233ebc0af19475e85aacf5fd41f1f99\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/2f69a488a83a42079fd3372f8a809d41 : SUCCESS in 2m 14s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/73a15005d89444fd98f0d6b9f518fcc2 : SUCCESS in 2m 27s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/bb67a1bbb9bc46f6b2589c6d14e86cbd : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":2},{"id":"3cddf15431c96735174b7c2bbbc0b7a0f88d155e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2023-09-25 02:01:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6829d1f275f1724d78932921d8631976b2357777","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 02:38:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb119beaafa34cf2905792cdf0aad3db\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/64a65914b5e8489fb77f779b03e963e8 : SUCCESS in 2m 49s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/038a90554c044b0c83bda854da470bfa : SUCCESS in 4m 36s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/c81ac0fed45b4de39c410c3400dac513 : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":3},{"id":"bb670277262d98d8bfb62115bc3ad98adfee9873","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-10-03 12:11:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c99e3b3d82f5f2b612caf98be3ccebe8014fc8a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 12:11:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"07a9c6fdd24eadb55dc64ad2449b12927f76ceb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 12:19:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f0f1860a2e347feb88f1b6e1b44ff24\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/ed4b6782253045cba79c5e6161f67e82 : SUCCESS in 2m 49s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d8514c4065da4be397e4999cbd716ea7 : SUCCESS in 2m 45s\n- ptgbot-upload-opendev-image https://zuul.opendev.org/t/openstack/build/639fa8feaaf145a797547c7f04e4ece3 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":3},{"id":"4edd92b7b06b730da52b9ad219b0b27b48269f77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 12:19:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"798933cec261efe8454da28a21c4002f39b5d28d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 12:20:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5671cac98910402da181041b850b2288\n\n- ptgbot-promote-opendev-image https://zuul.opendev.org/t/openstack/build/d913d35295684f85b15c5efe22bc6720 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0954e95fdcfdcf55d54752af77c33cc0f706c531","revisions":{"28759ca12f4bb91e0a08c3994500a74692bafedc":{"kind":"REWORK","_number":1,"created":"2023-06-07 14:47:57.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/02/885502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/02/885502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/1"}}},"commit":{"parents":[{"commit":"ee655d584940dd25bd7b41dc7b9d8b9509667fee","subject":"[docs]  Small fix for running http.server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/ee655d584940dd25bd7b41dc7b9d8b9509667fee"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-06-07 14:28:10.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-06-07 14:47:36.000000000","tz":-300},"subject":"[scripts]  Add an example/helper script for adding realtime fields.","message":"[scripts]  Add an example/helper script for adding realtime fields.\n\nChange-Id: I2accbc0db25d2e2d0700aad27b4e31de7ba63628\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/28759ca12f4bb91e0a08c3994500a74692bafedc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/28759ca12f4bb91e0a08c3994500a74692bafedc"}]},"branch":"refs/heads/master"},"2562cfacf432d4c5fad0c4d6a85fea5c3ca464b7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-25 00:21:22.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/02/885502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/02/885502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/2"}}},"commit":{"parents":[{"commit":"05ccf45dca90005ab16a2e3468f38b8a1b27c3ad","subject":"[docs]  Small fix for running http.server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/05ccf45dca90005ab16a2e3468f38b8a1b27c3ad"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-06-07 14:28:10.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-09-25 00:21:12.000000000","tz":600},"subject":"[scripts]  Add an example/helper script for adding realtime fields.","message":"[scripts]  Add an example/helper script for adding realtime fields.\n\nChange-Id: I2accbc0db25d2e2d0700aad27b4e31de7ba63628\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/2562cfacf432d4c5fad0c4d6a85fea5c3ca464b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/2562cfacf432d4c5fad0c4d6a85fea5c3ca464b7"}]},"branch":"refs/heads/master"},"0954e95fdcfdcf55d54752af77c33cc0f706c531":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-25 02:01:28.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/02/885502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/02/885502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/02/885502/3"}}},"commit":{"parents":[{"commit":"f3060c080671445bf3b432c7f820886089a314c7","subject":"[iCal] Add an iCal file route.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/f3060c080671445bf3b432c7f820886089a314c7"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-06-07 14:28:10.000000000","tz":-300},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2023-09-25 01:58:42.000000000","tz":600},"subject":"[scripts]  Add an example/helper script for adding realtime fields.","message":"[scripts]  Add an example/helper script for adding realtime fields.\n\nChange-Id: I2accbc0db25d2e2d0700aad27b4e31de7ba63628\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/0954e95fdcfdcf55d54752af77c33cc0f706c531"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/0954e95fdcfdcf55d54752af77c33cc0f706c531"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
