)]}'
{"id":"openstack%2Fptgbot~895835","triplet_id":"openstack%2Fptgbot~master~Ifbb7dcf2bf5f0b7ea4f53e26dd4bf4911c9b8d37","project":"openstack/ptgbot","branch":"master","topic":"bookworm-python3.11","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2023-09-20 15:53:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifbb7dcf2bf5f0b7ea4f53e26dd4bf4911c9b8d37","subject":"Update ptgbot to python3.11 and Bookworm","status":"MERGED","created":"2023-09-19 17:43:24.000000000","updated":"2023-09-20 15:55:04.000000000","submitted":"2023-09-20 15:53:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895835-bookworm-python3.11","meta_rev_id":"e3df4131ded15ae168b0e65eb1bc94d2eb19003f","_number":895835,"virtual_id_number":895835,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-20 15:53:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-20 15:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-20 15:43:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-19 17:53:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-20 15:21:23.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2023-09-20 15:43:21.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"06d96eb6f55d2eb5f21dfbf121ed82a7b217667d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-09-19 17:43:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8a2c5da2ba81dda1049d430fc292115861eebec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 17:53:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0d1514b97cb4d5fbb90d8076aacc401\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/cb4371ae705f4400b98bde310f40492e : SUCCESS in 2m 31s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/ed51f2d68b2f45e4a9c20717dc5e285b : FAILURE in 2m 33s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/36bd1eb58a1341dcb12d86db83b6ab03 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"12e9b6ad722f316c286fd890be08b86a88588ce2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-09-19 18:01:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"05d2d64f46415cf81e796f1f9426fc3bc37c16ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 18:07:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/164dda9387634396a942ab4736e2cda3\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c522b0f5d747daa0d204dfa1f66c5d : SUCCESS in 2m 26s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d1e4ccb380b242bdbf5b921a7668f749 : FAILURE in 3m 45s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/b91253610ef84ecf8db008f0b9be000e : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":2},{"id":"1f42d9c87c89e35e58509d5c46c8ef85d227efeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-09-19 18:14:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d21f2219d012d0eed1a1e28d52f69bd796861a21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 18:25:45.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f964e2c47104bfdb2f2d234e2560a0d\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/3edbd2ac1f204ed9836714c6203d2743 : SUCCESS in 2m 31s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/85f75f43c3194f75a10bbd6ffacebd6e : SUCCESS in 2m 35s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/b6276751d2dd4a9a81730206ac0d9a49 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":3},{"id":"fadbb78f96fd8647162c3124c448cb33269bf3a0","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-20 15:21:23.000000000","message":"Patch Set 3:\n\nWhy did test_message_process.py need those returned strings updated? Change in py311 behavior?","accounts_in_message":[],"_revision_number":3},{"id":"eebcf3f20de19b18d418bad384fac9d508cbbad7","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-09-20 15:35:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"13585f27db7d5ff878f1769c887f7d76072fdb02","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2023-09-20 15:38:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a71b079704a5bfa1d38c4016e25e5d7d0c794681","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2023-09-20 15:43:21.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"32cec8b1c174345fb4cfd8689767f7cbe892e57e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 15:43:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2a6ac5dbbcc48dda231bd2b5bfab80c9a8c4a668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 15:53:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0463542890474ee88f17c0ea57f78b5d\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/d52d39f733b24152980ab293847e3877 : SUCCESS in 2m 33s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/cb9b84127be848bcb0c0fea830a20838 : SUCCESS in 3m 55s\n- ptgbot-upload-opendev-image https://zuul.opendev.org/t/openstack/build/8a196d8f1f2f4f03bec3dd69746a257e : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":3},{"id":"7602b8cd363f603aec2cc830050cd27b0ca9cfdd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 15:53:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"e3df4131ded15ae168b0e65eb1bc94d2eb19003f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 15:55:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67529853ccfd49428446d86835e51cf5\n\n- ptgbot-promote-opendev-image https://zuul.opendev.org/t/openstack/build/f4a26d9342f141bba610fe47223b28a6 : SUCCESS in 30s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"513944afea7aa61110a2d8b30bcf3fc7bdb85c2a","revisions":{"3aba535bf77088fdb3b19d6d9d1f13fc9185e93c":{"kind":"REWORK","_number":1,"created":"2023-09-19 17:43:24.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/35/895835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/35/895835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/1"}}},"commit":{"parents":[{"commit":"70838b9d9328730ee6740838793d070cdaaf278c","subject":"Revert \"Migrate ptgbot to quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/70838b9d9328730ee6740838793d070cdaaf278c"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-09-19 17:42:21.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-09-19 17:42:21.000000000","tz":-420},"subject":"Update ptgbot to python3.11 and Bookworm","message":"Update ptgbot to python3.11 and Bookworm\n\nWe update python testing from python3.8 to python3.11 to match the\ndeployment setup in the Docker image. We then update the docker image\nfrom python3.9 + bullseye to python3.11 + bookworm. This is future\nproofing, but also OpenDev would like to clean up old docker container\nimage builds for bullseye and older python.\n\nChange-Id: Ifbb7dcf2bf5f0b7ea4f53e26dd4bf4911c9b8d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/3aba535bf77088fdb3b19d6d9d1f13fc9185e93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/3aba535bf77088fdb3b19d6d9d1f13fc9185e93c"}]},"branch":"refs/heads/master"},"f9d43e2a40984ea6f5ee92e46368a2b699d83a8f":{"kind":"REWORK","_number":2,"created":"2023-09-19 18:01:16.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/35/895835/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/35/895835/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/2"}}},"commit":{"parents":[{"commit":"70838b9d9328730ee6740838793d070cdaaf278c","subject":"Revert \"Migrate ptgbot to quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/70838b9d9328730ee6740838793d070cdaaf278c"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-09-19 17:42:21.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-09-19 18:01:10.000000000","tz":-420},"subject":"Update ptgbot to python3.11 and Bookworm","message":"Update ptgbot to python3.11 and Bookworm\n\nWe update python testing from python3.8 to python3.11 to match the\ndeployment setup in the Docker image. We then update the docker image\nfrom python3.9 + bullseye to python3.11 + bookworm. This is future\nproofing, but also OpenDev would like to clean up old docker container\nimage builds for bullseye and older python.\n\nChange-Id: Ifbb7dcf2bf5f0b7ea4f53e26dd4bf4911c9b8d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/f9d43e2a40984ea6f5ee92e46368a2b699d83a8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/f9d43e2a40984ea6f5ee92e46368a2b699d83a8f"}]},"branch":"refs/heads/master"},"513944afea7aa61110a2d8b30bcf3fc7bdb85c2a":{"kind":"REWORK","_number":3,"created":"2023-09-19 18:14:02.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/35/895835/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/35/895835/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/35/895835/3"}}},"commit":{"parents":[{"commit":"70838b9d9328730ee6740838793d070cdaaf278c","subject":"Revert \"Migrate ptgbot to quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/70838b9d9328730ee6740838793d070cdaaf278c"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-09-19 17:42:21.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2023-09-19 18:13:56.000000000","tz":-420},"subject":"Update ptgbot to python3.11 and Bookworm","message":"Update ptgbot to python3.11 and Bookworm\n\nWe update python testing from python3.8 to python3.11 to match the\ndeployment setup in the Docker image. We then update the docker image\nfrom python3.9 + bullseye to python3.11 + bookworm. This is future\nproofing, but also OpenDev would like to clean up old docker container\nimage builds for bullseye and older python.\n\nChange-Id: Ifbb7dcf2bf5f0b7ea4f53e26dd4bf4911c9b8d37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/513944afea7aa61110a2d8b30bcf3fc7bdb85c2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/513944afea7aa61110a2d8b30bcf3fc7bdb85c2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
