)]}'
{"id":"openstack%2Fptgbot~956718","triplet_id":"openstack%2Fptgbot~master~I634ddea64366b61275d636b3e91f9d6d11ad5213","project":"openstack/ptgbot","branch":"master","topic":"move-to-quay","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2025-08-06 21:20:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I634ddea64366b61275d636b3e91f9d6d11ad5213","subject":"Reapply \"Migrate ptgbot to quay.io\"","status":"MERGED","created":"2025-08-06 17:48:54.000000000","updated":"2025-08-06 21:21:28.000000000","submitted":"2025-08-06 21:20:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956718-move-to-quay","meta_rev_id":"182e7a3cdf0b3ab2b13699723164bfebaf4913fe","_number":956718,"virtual_id_number":956718,"owner":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 21:20:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-06 19:10:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-06 21:13:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 18:06:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-06 19:10:09.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"778302053e49ef16935487adbae353abadc42c0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-06 17:48:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fad0dca9023228accc17a24fa1b899ff4ebc7e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 18:06:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9502e0bdb3b47e2bbcb7fecfda18dd3\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/d42a324eec26499d84df1e6f4f326e19 : FAILURE in 2m 30s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/9eb6735040884af0ab0c1312455bd739 : SUCCESS in 4m 30s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/b849e8a3c2c5411e899c9d9a10f819a9 : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":1},{"id":"54d17d7cf25b05ea7ce79ab51a45b38b4a59eea3","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2025-08-06 18:33:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"827246e8e3cb07742dbc62297ac558f2ed330379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 18:39:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5363252f3fb4d239b6fbe0db7ce4f66\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/976db793748c450ebb711ff0b83a9257 : SUCCESS in 4m 04s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/d281ff86ccc34c31bf326dde9138b9ba : SUCCESS in 3m 04s\n- ptgbot-build-opendev-image https://zuul.opendev.org/t/openstack/build/801c13ab224c4cafaab4602a9417c36f : SUCCESS in 5m 00s","accounts_in_message":[],"_revision_number":1},{"id":"eb7e68fff7a20874064726c32b2579e08b54d20f","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-08-06 19:10:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"62fba69891f31ad62d3fa5b8e8d847fd3e359356","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-08-06 21:13:02.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b016c14cf8c38c1aa2f1b5d90472c6962786546c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 21:13:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e02dcefa3bf2dd484f27adf49e6696f5cc830f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 21:20:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ba28518289415e8adc9773b3b60e93\n\n- tox-pep8 https://zuul.opendev.org/t/openstack/build/c4296ed4d1c045f091f2b11da27a6b69 : SUCCESS in 1m 33s\n- tox-py311 https://zuul.opendev.org/t/openstack/build/09e7ec87831a48608d28097a3e3a87c7 : SUCCESS in 2m 14s\n- ptgbot-upload-opendev-image https://zuul.opendev.org/t/openstack/build/2e4aa1b2b95048e793193d2c4f418f60 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b51fc381c93f85b275c278a96bbdd16d18e7b7bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 21:20:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"182e7a3cdf0b3ab2b13699723164bfebaf4913fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 21:21:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32ab0c17926b4a05abca5d0513f0fb6c\n\n- ptgbot-promote-opendev-image https://zuul.opendev.org/t/openstack/build/a9ed25cb45574ef39fe43ea49bb77cbf : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a4c2ef0a24962e6c254a4d689f0a29a95f53c62","revisions":{"6a4c2ef0a24962e6c254a4d689f0a29a95f53c62":{"kind":"REWORK","_number":1,"created":"2025-08-06 17:48:54.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/18/956718/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ptgbot","ref":"refs/changes/18/956718/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/18/956718/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/18/956718/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ptgbot refs/changes/18/956718/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ptgbot refs/changes/18/956718/1"}}},"commit":{"parents":[{"commit":"287f8d50636e829f2ec9d482238bf25d39d40dc1","subject":"Set SO_REUSEADDR and SO_REUSEPORT to enable faster service restarts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/287f8d50636e829f2ec9d482238bf25d39d40dc1"}]}],"author":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-08-06 17:47:38.000000000","tz":-420},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2025-08-06 17:48:37.000000000","tz":-420},"subject":"Reapply \"Migrate ptgbot to quay.io\"","message":"Reapply \"Migrate ptgbot to quay.io\"\n\nThis reverts commit 70838b9d9328730ee6740838793d070cdaaf278c.\n\nNow that eavesdrop is running Ubuntu Noble and podman we can host the\nimage on quay without losing speculative image testing. Do so.\n\nChange-Id: I634ddea64366b61275d636b3e91f9d6d11ad5213\nSigned-off-by: Clark Boylan \u003cclark.boylan@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/6a4c2ef0a24962e6c254a4d689f0a29a95f53c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ptgbot/commit/6a4c2ef0a24962e6c254a4d689f0a29a95f53c62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
